)]}'
{"id":"openstack%2Fmistral~725102","triplet_id":"openstack%2Fmistral~master~I3f1e37c43c6d87a678e0ffaa0fcbd607a27dbc7b","project":"openstack/mistral","branch":"master","topic":"rally_big_wf_scenario","hashtags":[],"change_id":"I3f1e37c43c6d87a678e0ffaa0fcbd607a27dbc7b","subject":"added a new realcase scenario to rally and clean leftovers","status":"MERGED","created":"2020-05-03 13:32:11.000000000","updated":"2020-05-09 18:00:25.000000000","submitted":"2020-05-09 17:58:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725102-1589047134333-eaa21e4c","meta_rev_id":"afbbf2073f28cb1047e84f186ac6fc2987e09864","_number":725102,"virtual_id_number":725102,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-09 17:58:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-05-09 17:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-09 13:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-05-09 17:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 13:24:22.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-05-09 17:09:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-05-09 17:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7749aeff84272d51ecb7ff955ea343a99392ac6","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-05-03 13:32:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40977ab4ae1735022f931998a5671b6b35bc6034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 15:16:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6262d9bcb84590b00e3a69cb0085fc : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3a7653f3fa54c7b806f6c2fa8512806 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b390a473337847b3b0a00ccbd68c3755 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f37846848a480091455960e3511f04 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25d3080664d43258dfd368d32483ef3 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62e789cf12dd4eb69880194917c71269 : SUCCESS in 8m 33s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4de991574ab341ce9bc36e788ebb836f : SUCCESS in 41m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b931377275c43128a30e0a9a19cd07e : SUCCESS in 40m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c222d16212994081b92e9e83105335f2 : SUCCESS in 43m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c63239566f5b4aa395c15b90876ea702 : SUCCESS in 45m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/81618f8b9a3d4cf6aa77efe353d503ff : SUCCESS in 41m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/08b41511e14a40dca7eb3d626669ad50 : SUCCESS in 30m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a1e4925968b344539d33539246874208 : SUCCESS in 28m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/73a9fd4235e548929a9013191a5683a1 : FAILURE in 1h 39m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70f73a528e894681a17a39f95a361381 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7dc9962ca4cb51d6d89796033f18ec0522a49349","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-05-04 10:01:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64aeb45dd0dd0abb12ca7d709f77a506c9ef7f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 10:54:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994b540857514e62b82c9966ae137fa4 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65f9b2f5bb714edcb4efed1ed60bd61f : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe0749bee4a94a8281b72640a2cf9ffb : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc4f4b0df1240a19c927867f254a4ae : SUCCESS in 11m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7487bfdc3d24201a31d17ac5b777a99 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8155c3217154f599a0b20c92fd96836 : SUCCESS in 9m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d3b30fe9566543adbbe902bc3ca77494 : SUCCESS in 40m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/101efddcbef74a70b052804da4443711 : FAILURE in 7m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2a4e4ac173d47658d07040012009f20 : SUCCESS in 48m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d258e002ad1f4b97bd9ba136d92e3e4a : SUCCESS in 37m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/85ec2c87d4cc45499898f429282153ca : SUCCESS in 41m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/40b3d8396a404d7789cc650a31e35f07 : SUCCESS in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/71050b154c234b7397354fcf52ddc363 : SUCCESS in 28m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5d38ab02d3a244b0a629545fd6619396 : FAILURE in 7m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/153cba016d41468ea284f089c6563cfc : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":2},{"id":"df17b5b8dc040144bfe2c47dbdbc8f657e222cf1","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-05-04 11:53:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4124a3390febeb37915430f4c6e8c85091245cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:37:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90e521281504c3881d987dbe3d3ae4f : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad4db60c8fe9482eb79a1b00836185cf : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63cf6497cd634c499d818032bf91e0dc : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09034ccc86914f5b80a1d71e79513a25 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c000ca9660346368e91ddcee25cc77c : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e05bede7be3c4d3d887367b890ea422e : SUCCESS in 8m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/69df1d8b4fae4a68a0b1d1b651a07f77 : SUCCESS in 44m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3b46f3b948b4b33b8f640be925d6b8a : SUCCESS in 41m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a880d4872426469f937d0e20f6effae0 : SUCCESS in 48m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c44cc05390ae485a9cf0ce27d116ff20 : SUCCESS in 41m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1dd5a744d05e4b5cb75b1f78a8f6bc2d : SUCCESS in 43m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9a259905ec4418582b4be14d104d018 : SUCCESS in 31m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7fc7caf077e946a69184f4e7bffd57f4 : SUCCESS in 30m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8d97d7e555e5433a9412e63d813c8a64 : SUCCESS in 1h 39m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/868c0c9e152646a7b02a22438fb47ab7 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"000033c3772831b1fd56154e796fe15bd1f0558b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-05-05 10:27:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d81bc0efbdb0155001ff2345bc3efefba9f306a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:11:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27a43adacf7481db857920e872efa35 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb71efc23ca401580b7536d848bc168 : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92b8890330b44e13836274ab89583e01 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e79bd04879746f2b3cb8ebc203f31aa : SUCCESS in 14m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c066a2d326742abbdda98cd14c4eae0 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/049991aa6be847269a55b99e432cf92a : SUCCESS in 9m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1464d280180443e3a4106b2d3fae85e4 : FAILURE in 36m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b293405829a3475fad46fb0f76efede2 : FAILURE in 46m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56e4aa93275e41ec8d24c518a0ae8288 : FAILURE in 42m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6f51b30098984bc3ac353e9c7f9bbe15 : FAILURE in 42m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3a25e40680a042a4bd167b01dab45c04 : FAILURE in 44m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c84895a66f2b44719b0f4b064fd7b4f8 : SUCCESS in 29m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0facc75c266d44e4b92ee5ebc0250f0b : SUCCESS in 30m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c8e0a7cd3bff43dcad12aea666afde4d : SUCCESS in 1h 41m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0ee64b421ee47e8b7bc7da616351576 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"51af5e95218b5a1e2e0984644e15f7df1c9bd633","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-05-06 06:25:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2061db932685720a380d9c4ad51c169b65c1a2e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 08:08:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0914bbc338d74e43a4cf58e02c668554 : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82713764c0404df5a6aebc26defa78ee : SUCCESS in 15m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b459656b0679414ca1926a6adc8cf1ae : SUCCESS in 12m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8a9f83c45c4284b950298dae852032 : SUCCESS in 11m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec43c2e83ad54e9a93b0748d906a76fe : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1f46bf2316e46f3836073d8c7063630 : SUCCESS in 12m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/59d738f8a4324c1ca8bc7bdb2217908d : FAILURE in 32m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbbfe27e5f14271b0db82b711ce89f1 : FAILURE in 32m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d878b2a4be624f299d38e7be961ac4bd : FAILURE in 39m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/51806d06c3c34087bbe1eaabcffe2e44 : FAILURE in 50m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5cd789d9d28c4a81adea224f49bfc40e : FAILURE in 38m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/125dbabf80e64842ab3415532cae8473 : SUCCESS in 32m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/93cce7da2d88432984889837e4a42fe4 : SUCCESS in 35m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6930dfdaf82f4651bf66be053ff615a0 : SUCCESS in 1h 36m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/709c73bb325f42969fcb62374e47863f : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f9af3d4e20ffb58281418c0c504989b3fa9b3870","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-07 05:48:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3f9b4ff83192d0cda67761e5a1516aa9a5bea12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 07:35:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac82ed1e66f4483ad5f1fb184378b1f : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68d76a039a9400797f855c979495c1f : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a1687ef37443b5802ba39ae85a9695 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc713e22a2954c32a1370f11fc2299ae : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41ca7d0d04764a0c9c675c0ec79799da : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b61fa596cf743a2b0ce32144b4d57fc : SUCCESS in 8m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ade5bb7025924bcc94e6757562646072 : FAILURE in 35m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90e9a54d73c64c7ba2d73eda60869464 : FAILURE in 36m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daeeff5ca7ca42eea2d0e014a53dfa8e : FAILURE in 38m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a0e794470645448e8d0c2c9bdb03143f : FAILURE in 37m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b3b5766c9a454af2a8396978782e74d9 : FAILURE in 39m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f5af13d3ff6e4f33a35a425dbc168416 : FAILURE in 30m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/64301c74af5147b7958d296a80b6dd00 : SUCCESS in 28m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e4f6fe37a0db444ca399a39ff6343ef7 : SUCCESS in 1h 41m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e41dc2d22ac4e16b802ed2bc9ee8f16 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9a1d205122d431e0de34751dd1932d1c9d50bce1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-07 20:29:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6c231915e41bb942e84c566fd5ab6121091067b1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-07 21:15:34.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a3fe6eaafe072d95d99f82dbf56a8dffa434046e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 23:07:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d701e56d73640b3b440cd7435e81fa9 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b1858010f246bd84ad0b67c4c404e1 : SUCCESS in 9m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2472eae9838f47c8befd6287fc174441 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43d9f057e3d7479c9f3cd2eec357bbbb : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cf21793f7c04a78a7b92b8dca9cb807 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8768a3916fbd4fd5a41c581f91b7c527 : SUCCESS in 9m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/eca2c4c9db954121ba6acb7b503420c7 : FAILURE in 46m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2c60bfc2baa4ce5940c5fe82d859c4a : FAILURE in 46m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed4978ac583e4227bd14f30fe9dcf6fe : FAILURE in 35m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dffda4294fad44099ca41f38be631a98 : FAILURE in 36m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/296aeb7c3c31477e9fda913816256cad : FAILURE in 35m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0e5a0c20b8cb415fb129ece3e0e44c82 : SUCCESS in 31m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f43e24edab204f93a53fb09b9e95c9a7 : SUCCESS in 29m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c98a4771cdd348b780bd99d0cce6e5f2 : SUCCESS in 1h 50m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b988f0d86604ade98b16f3f55557631 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":4},{"id":"08e135245ab16dc4cdd0a569c7bedc6b3433df63","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-08 05:24:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"80c481f82e3b7a9e4437b148aba64b059ab7c600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 07:09:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce68501552564befb98ab6f862c6b8d2 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4edf1c1d27543319217b0a822da67ac : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/069e9dea3e63448d84590b5f54ce075a : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a96c1085ac485a85bac221169aa92f : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa560e78800048bfbe5f398fcdcad1f8 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5ad4790739247e8b2fa562b37549b72 : SUCCESS in 9m 16s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e180c37ab2ed45a0ac2f81f16a8df992 : FAILURE in 38m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd795c9be0514a90a3cb3e471284f309 : FAILURE in 36m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddb7d3d791bb41ba987da77cd9d36cf0 : FAILURE in 34m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/df5167eb261e4dc285b14f0f32dbede9 : FAILURE in 34m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e3b81112470c4022b7cea9b5fa66ebcc : FAILURE in 35m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0583d5d7679e4f02a9f592841ec3f718 : SUCCESS in 29m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0dd0113ff6df40aa93efb833885e32f1 : SUCCESS in 27m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c83034fe01ec4f9cad71d05a2568d21d : FAILURE in 1h 44m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fdf3934e64c4182887184b32ebc7da5 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":4},{"id":"2bf8f12b387cc07a90c43c84a6a318c35c61b3a7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-08 12:48:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f546d918558659550af5957f3f89a7162bad3ebb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-08 13:31:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"706aa0236c32fd4c69932c1232a5e2d1db601163","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-08 13:31:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"c953e887921025d31b8ebf84008fbdba6a7d8e0f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-08 14:25:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"f13b8e871d0e755137645bfa44386db39ce86120","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-08 14:25:24.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"6c6ac033703899679d8d3619d3b7458c0fa25816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 16:23:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d19f8eab8794804b527ca998cb65781 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/915aa4181b594bf283f4893084d9810a : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/378b83003b7e463cb83d22f49ec59be1 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc931a1a0824ac9a5b691314bbd9e76 : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d17342895ac1491c89df991b7b8c007d : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01ad53bc7d634f609897c993009e010f : SUCCESS in 9m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ca827e0f73ac45cd953becae179a5442 : SUCCESS in 48m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4401d12fe8b64a1bb53565b7a038aa2b : SUCCESS in 50m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2600bc6781f41c2ba58bf09557e2f70 : SUCCESS in 43m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c9a9828daedd420ca5453a789d2d35a3 : SUCCESS in 50m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/54971d18ac494e2396a04e6b1f02c464 : SUCCESS in 46m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ee6ad9ee6e314f52b5921a8165c805f4 : SUCCESS in 29m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e7e43c0ebfe34f45b79939f9cba7d838 : SUCCESS in 27m 40s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/635c57f6e84c402797301ad4afdfb4b8 : SUCCESS in 1h 50m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41b4bae470524f27b9421bacaf5d020f : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":4},{"id":"19e4a22bef7b11b9fa017be1f9b1cf4c19a0a3ec","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-09 13:24:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9e8e86be114ebc617d791090ebd0c90831aa4937","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-09 17:09:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a71248fbb43262b2c7e91b588d3ccb24da9fdf33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 17:10:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a7961288ae49d29820d76b074a25b984dd7b79d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 17:58:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e66f4cf39f1c76b8621486579a22517794c4f10f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 17:58:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0684eec2d8b141ca843d35724d3c8b35 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da24bec319bf4e809c30c1d46bb99f23 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc09a1123809478981f041d93c98b0e0 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b169927ae14706ae1e0f616420373b : SUCCESS in 6m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c544ebe0a98648a2a6e7427fe3c5c89e : SUCCESS in 36m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5334abf0cb845728aedeb9cbcbbd844 : SUCCESS in 45m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77a29209e1864aaf89e68ec8d5791e30 : SUCCESS in 34m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b374bb88936d49a084af63c95a21f81d : SUCCESS in 41m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/993fefa7a6084195926b949da78390b4 : SUCCESS in 29m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f78b131d941c440f89c1dac51a4930f5 : SUCCESS in 27m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0f418e0cd4c94923918d349814612d7e : SUCCESS in 43m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56840ff5bdd24fdb99986ccbbfdf5233 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":4},{"id":"63df7053aa4a6f1b1f77b17d8a5fabd26c1a6258","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 18:00:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a72ecac42ab4dfcb2725eb53c68b935 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1f9e1c1f0abf6c2ce323f30ddebda95bddc0ae64","revisions":{"2afb641f80c2be329b681df5ac0442f23d289005":{"kind":"REWORK","_number":1,"created":"2020-05-03 13:32:11.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/02/725102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/725102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/725102/1"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-05-03 13:31:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-05-03 13:31:58.000000000","tz":0},"subject":"added a new realcase scenario to rally","message":"added a new realcase scenario to rally\n\n -added another new big workflow scenario (Terminate) to rally\n  scenarios, this will help us in the future benchmark mistral\n  and notice performance regressions if they happen.\n\nChange-Id: I3f1e37c43c6d87a678e0ffaa0fcbd607a27dbc7b\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2afb641f80c2be329b681df5ac0442f23d289005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2afb641f80c2be329b681df5ac0442f23d289005"}]},"branch":"refs/heads/master"},"a402528ce98ea79fd842be0dfeb2a5e9d203bce3":{"kind":"REWORK","_number":2,"created":"2020-05-04 10:01:36.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/02/725102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/725102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/725102/2"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-05-03 13:31:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-05-04 10:01:14.000000000","tz":0},"subject":"added a new realcase scenario to rally and clean leftovers","message":"added a new realcase scenario to rally and clean leftovers\n\n -added another new big workflow scenario (Terminate) to rally\n  scenarios, this will help us in the future benchmark mistral\n  and notice performance regressions if they happen.\n\n -delete leftovers (workflows,actions) after the scenario has\n  finished running.\n\nChange-Id: I3f1e37c43c6d87a678e0ffaa0fcbd607a27dbc7b\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a402528ce98ea79fd842be0dfeb2a5e9d203bce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a402528ce98ea79fd842be0dfeb2a5e9d203bce3"}]},"branch":"refs/heads/master"},"60f7007bfdc1ec15f7f65d6947fc0eae46b1e83e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-07 20:29:39.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/02/725102/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/725102/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/725102/3"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-05-03 13:31:58.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-05-07 20:29:36.000000000","tz":0},"subject":"added a new realcase scenario to rally and clean leftovers","message":"added a new realcase scenario to rally and clean leftovers\n\n -added another new big workflow scenario (Terminate) to rally\n  scenarios, this will help us in the future benchmark mistral\n  and notice performance regressions if they happen.\n\n -delete leftovers (workflows,actions) after the scenario has\n  finished running.\n\nDepends-On: https://review.opendev.org/#/c/726238/\nChange-Id: I3f1e37c43c6d87a678e0ffaa0fcbd607a27dbc7b\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60f7007bfdc1ec15f7f65d6947fc0eae46b1e83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60f7007bfdc1ec15f7f65d6947fc0eae46b1e83e"}]},"branch":"refs/heads/master"},"1f9e1c1f0abf6c2ce323f30ddebda95bddc0ae64":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-07 21:15:34.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/02/725102/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/725102/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/725102/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/725102/4"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-05-03 13:31:58.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-05-07 21:15:31.000000000","tz":0},"subject":"added a new realcase scenario to rally and clean leftovers","message":"added a new realcase scenario to rally and clean leftovers\n\n -added another new big workflow scenario (Terminate) to rally\n  scenarios, this will help us in the future benchmark mistral\n  and notice performance regressions if they happen.\n\n -delete leftovers (workflows,actions) after the scenario has\n  finished running.\n\n\nChange-Id: I3f1e37c43c6d87a678e0ffaa0fcbd607a27dbc7b\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f9e1c1f0abf6c2ce323f30ddebda95bddc0ae64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f9e1c1f0abf6c2ce323f30ddebda95bddc0ae64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
