)]}'
{"id":"openstack%2Fmistral~726197","triplet_id":"openstack%2Fmistral~master~I81fc508d8d34a165ad2cb2fc15928db795f163be","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I81fc508d8d34a165ad2cb2fc15928db795f163be","subject":"[WIP] Remove six","status":"ABANDONED","created":"2020-05-07 16:26:15.000000000","updated":"2025-02-15 17:59:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef9392e55890a1f6fd4cc244e027163bfb4b53e6","_number":726197,"virtual_id_number":726197,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-09 16:52:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":-1,"date":"2020-05-10 05:22:03.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 16:52:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8871911e3157da8ed34463c5a859027de51a2cde","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-07 16:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db5f8b230fc7ef58174820c541cceae94178ad92","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-07 16:32:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b5a0ef40dc101ef5dc455364091caa57f94442d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 18:43:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba03efbbbba54cfeabcfa0880a0c1eec : FAILURE in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3c33cbb8c0a4c49904c8df93c6cbb23 : FAILURE in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43ccbaf0b71147f0b1fbe75046a8167f : FAILURE in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9841c699f0e14c6e8dd7294d72609768 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ece131a764824bdf9a725572d684f0df : FAILURE in 7m 32s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebcb124eed4648258f870739a2ba279d : FAILURE in 10m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/145ffcd6ce65416bb13acd36662ff1aa : FAILURE in 37m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fe22ed2fa234338acfc55ee381b3d3c : FAILURE in 48m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8457bfe6ac247b7b2fa7bc555233838 : FAILURE in 45m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5af0c41a2f7749e2b03c5e41e33d2af6 : FAILURE in 50m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/46cc4bc2313e465a9731a01970065a0b : FAILURE in 46m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c1a271db834849a4a6b086d47c3bf1ee : FAILURE in 30m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5af15d062a9a4aebb52ba907c1edfe92 : FAILURE in 30m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fa5e42932baf4914b1b40f78da4f2261 : SUCCESS in 2h 05m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1ab123755f4ffe9ca7ca3f6b0da9e7 : FAILURE in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2ad56942713daf4d82c03c9ec3f6252d93b192f1","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-09 16:15:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"283e592cf9c488ab297cb566f3658799a35e84f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 16:52:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1302880be83545ba8210a2bb75139a9a : FAILURE in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22a2d8d3e33640029e3a5fdba7ec13e1 : FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4c59511da744f698683414c6a3a2c0d : FAILURE in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbac84534c9d4675b5f392009bfd3e49 : FAILURE in 6m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d227be40cbeb4b3cabc86e959b3f2c81 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f661421497040cfa944356eb304c72b : FAILURE in 5m 19s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eab2e7166f7c43aea6cd0eca0d158118 : FAILURE in 5m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/71b236c54c6a4c209a088cb6afda54bf : FAILURE in 30m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9ba9cbd2ae24f19b448d09447c1ac6a : FAILURE in 34m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ecf9fcd68164f2d925067a8709d8982 : FAILURE in 32m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/14286282caae4a9e90ca86d7970e2c9f : FAILURE in 34m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/edb69e7e2f004c31a85d5393b8ac3378 : FAILURE in 31m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fc12c4b3e0b6415e8791650d82d310d3 : FAILURE in 4m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/654723ba976d48d68df7f10623c0ff0d : FAILURE in 4m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/375da99c2ee8409ab0f73acae6261ac7 : FAILURE in 29m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f82f456dad124c1ba0d1f0c702c89059 : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"80c2450bcc816e9516344aa4dea8c4b3a747ad28","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-10 05:22:03.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ef9392e55890a1f6fd4cc244e027163bfb4b53e6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 17:59:04.000000000","message":"Abandoned\n\ndone in another change, cleaning","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9483a9b871dc229286ad67a139f4ae4da9b41d0","revisions":{"8c8087efa90daeae790fcc071ddea70147147091":{"kind":"REWORK","_number":1,"created":"2020-05-07 16:26:15.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/97/726197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/726197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/726197/1"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-07 16:21:44.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-07 16:25:50.000000000","tz":480},"subject":"[Part1] Remove six","message":"[Part1] Remove six\n\nRemove the six from tests\n\nWe don\u0027t need this in a Python 3-only world.\n\nChange-Id: I81fc508d8d34a165ad2cb2fc15928db795f163be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8c8087efa90daeae790fcc071ddea70147147091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8c8087efa90daeae790fcc071ddea70147147091"}]},"branch":"refs/heads/master"},"5c0a3aa4ad0cafb92e05c49a3ce627cef9c00693":{"kind":"REWORK","_number":2,"created":"2020-05-07 16:32:16.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/97/726197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/726197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/726197/2"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-07 16:21:44.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-07 16:31:31.000000000","tz":480},"subject":"[Part1] Remove six","message":"[Part1] Remove six\n\nRemove the six from tests\n\nWe don\u0027t need this in a Python 3-only world.\n\nChange-Id: I81fc508d8d34a165ad2cb2fc15928db795f163be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c0a3aa4ad0cafb92e05c49a3ce627cef9c00693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c0a3aa4ad0cafb92e05c49a3ce627cef9c00693"}]},"branch":"refs/heads/master"},"f9483a9b871dc229286ad67a139f4ae4da9b41d0":{"kind":"REWORK","_number":3,"created":"2020-05-09 16:15:58.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/97/726197/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/726197/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/726197/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/726197/3"}}},"commit":{"parents":[{"commit":"f57d49f52cc539cf8314a9e62eced8373a76c3e1","subject":"Merge \"Add a Rally scenario for a very big mistral workflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f57d49f52cc539cf8314a9e62eced8373a76c3e1"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-07 16:21:44.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-09 16:14:20.000000000","tz":480},"subject":"[WIP] Remove six","message":"[WIP] Remove six\n\nRemove the six from tests\n\nWe don\u0027t need this in a Python 3-only world.\n\nChange-Id: I81fc508d8d34a165ad2cb2fc15928db795f163be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f9483a9b871dc229286ad67a139f4ae4da9b41d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f9483a9b871dc229286ad67a139f4ae4da9b41d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
