)]}'
{"id":"openstack%2Fmistral~727122","triplet_id":"openstack%2Fmistral~master~I6781ad6f8259ff1af0adbf2c0a9574db5272360c","project":"openstack/mistral","branch":"master","topic":"bug/1878167","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2026-03-28 19:39:48.000000000","reason":"Change was abandoned"},"8731":{"account":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"last_update":"2026-03-28 19:39:48.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-06-03 10:16:14.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I6781ad6f8259ff1af0adbf2c0a9574db5272360c","subject":"WIP: fix premature evaluation of \"publish\" in case of \"retry\" policy","status":"ABANDONED","created":"2020-05-12 09:10:18.000000000","updated":"2026-03-28 19:39:48.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"996a4a58ab43754d31c92c22f6ba52abf6c02b27","_number":727122,"virtual_id_number":727122,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-12 13:17:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":-1,"date":"2020-05-26 08:53:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}],"CC":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 09:11:00.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2020-05-12 09:13:12.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-05-12 13:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 08:13:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-09 08:13:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"},{"updated":"2025-06-03 07:20:01.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"CC"}],"messages":[{"id":"4bfbdd0fcee897dbe301830e527dcb7fb3ab38fb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-12 09:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"840715b1b80e530600a61c587aefe7d6ad1c992f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-12 09:12:47.000000000","message":"Patch Set 1:\n\nFor now it\u0027s just a unit test that reproduces the problem. I described the issue itself in details in the ticket: https://bugs.launchpad.net/mistral/+bug/1878167","accounts_in_message":[],"_revision_number":1},{"id":"3d525b6037bece3f094122b7e9fd3cd458bf533e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:08:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02efee81c22245baaf2eb1c64056ac35 : FAILURE in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56826e05549c49f4a7bad13f4be02071 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67fedc7b70a541f2a5126010c5617803 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65019696e36843eb884b848d4d85d4f8 : SUCCESS in 8m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baab8060f40e41efa6ffb1470c3cf992 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b223fab24412456885b017c88ed373ef : SUCCESS in 10m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6f54862a6fce44868aa3a1d672c3acf1 : SUCCESS in 42m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/581d4a8379444cfcbc5299ab0c411746 : SUCCESS in 42m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/316072c9b94a48fca350c0fc116959b3 : SUCCESS in 39m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9fcde3b113b94683bf1a5c491ce04e3a : SUCCESS in 42m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/928bf8f2c8054885abce4f36993639a0 : SUCCESS in 43m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/63ba31fb9a3e4a378f1de5a45de94e53 : FAILURE in 31m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/364fe7f242f348bcb08a53a50b05ab38 : SUCCESS in 31m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ea3f930ed8984aafbfb56fee3208df03 : SUCCESS in 1h 49m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b93f350681f142a89a25e6266528134e : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"287c1ee8196befcf41292e0f602f01fd5cd6a6d5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-12 11:41:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0bf71ac1ca0187f229497ff2525541037eae3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 13:17:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3a0b4ed6214713849a8d1d0c90fd6b : FAILURE in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df2e20b5da7140f9973fe2db0007a212 : FAILURE in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b833fc2cf35044bb9d6e9c593308467c : FAILURE in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5085d60ddd4240adca4722bbbd4eec : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11ecc80c28bc41a59dd3d4889c19f92e : FAILURE in 10m 01s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47a07387894147d7b2efa8e9c282dce1 : FAILURE in 9m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/521de61014fc4c948e80fcfa203852a7 : SUCCESS in 39m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93afbe7b4b6540c5b68483d8e5f83e0a : SUCCESS in 46m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64576465e60747caa77ef19d77299d8e : SUCCESS in 42m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c988464ae01482da32fd8163693cbd0 : SUCCESS in 38m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8e5712797e674936987f745382ba4d4a : SUCCESS in 38m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f2888dc900af4b8597ab2498a06ebd46 : FAILURE in 32m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0c8eeae7490e476caafaa0572b29dcc5 : FAILURE in 30m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5d4493b0e5154362aac61142beb48911 : SUCCESS in 1h 34m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/274895ec37674313b9eee2e8ccef6bce : FAILURE in 9m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b04d55048935f224a2abd2165d4fdf816070558c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-26 08:53:14.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"0afbae26d205ffc54f4d06fc0bc3006ba2c3f231","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-09 08:13:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d49c503b5304f4616c0e678f94e894398b7ff3a0","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2025-06-03 07:20:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fdabb32a4f6706b5a8b84915042a69569276d6ce","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-03 10:16:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"996a4a58ab43754d31c92c22f6ba52abf6c02b27","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:39:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e57af2332c155495c7ab23cdc8f4dfaf5481f0d2","revisions":{"567a149dcbaba647e85e09ee9435eebb4df18e64":{"kind":"REWORK","_number":1,"created":"2020-05-12 09:10:18.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/727122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/727122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/727122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/727122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/727122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/727122/1"}}},"commit":{"parents":[{"commit":"90bcceedef43b6484f0598344e30f07712dca8f0","subject":"Merge \"added a new realcase scenario to rally and clean leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90bcceedef43b6484f0598344e30f07712dca8f0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-12 09:07:16.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-12 09:08:27.000000000","tz":420},"subject":"WIP: fix premature evaluation of \"publish\" in case of \"retry\" policy","message":"WIP: fix premature evaluation of \"publish\" in case of \"retry\" policy\n\nTODO: Change the design task policies so that \"retry\" has a chance\nto change the flow even if the task action is successful.\n\n* Added a unit test\n\nChange-Id: I6781ad6f8259ff1af0adbf2c0a9574db5272360c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/567a149dcbaba647e85e09ee9435eebb4df18e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/567a149dcbaba647e85e09ee9435eebb4df18e64"}]},"branch":"refs/heads/master"},"e57af2332c155495c7ab23cdc8f4dfaf5481f0d2":{"kind":"REWORK","_number":2,"created":"2020-05-12 11:41:18.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/727122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/727122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/727122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/727122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/727122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/727122/2"}}},"commit":{"parents":[{"commit":"90bcceedef43b6484f0598344e30f07712dca8f0","subject":"Merge \"added a new realcase scenario to rally and clean leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90bcceedef43b6484f0598344e30f07712dca8f0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-12 09:07:16.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-12 11:40:12.000000000","tz":420},"subject":"WIP: fix premature evaluation of \"publish\" in case of \"retry\" policy","message":"WIP: fix premature evaluation of \"publish\" in case of \"retry\" policy\n\nTODO: Change the design task policies so that \"retry\" has a chance\nto change the flow even if the task action is successful.\n\n* Added a unit test\n\nChange-Id: I6781ad6f8259ff1af0adbf2c0a9574db5272360c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e57af2332c155495c7ab23cdc8f4dfaf5481f0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e57af2332c155495c7ab23cdc8f4dfaf5481f0d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
