)]}'
{"id":"openstack%2Fmistral~731491","triplet_id":"openstack%2Fmistral~master~Ie728bf950c4244db3fec0f3dadd5e195ad42081d","project":"openstack/mistral","branch":"master","topic":"refactor_policies","hashtags":[],"change_id":"Ie728bf950c4244db3fec0f3dadd5e195ad42081d","subject":"Refactor task policies","status":"MERGED","created":"2020-05-28 10:37:32.000000000","updated":"2020-06-02 10:58:28.000000000","submitted":"2020-06-02 10:56:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731491-1591095415018-b17e9e7b","meta_rev_id":"fb5679e62c47ff53febe170df20bdb64a61814df","_number":731491,"virtual_id_number":731491,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},{"value":0,"date":"2020-06-02 08:52:03.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 10:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-06-02 07:02:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2020-06-01 13:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-01 08:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-06-02 07:02:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-06-01 09:19:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-06-01 09:19:21.000000000","_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 08:57:23.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-06-01 09:19:14.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-06-01 09:19:21.000000000","updated_by":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"reviewer":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"state":"REVIEWER"},{"updated":"2020-06-01 13:46:17.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-06-02 07:02:37.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-06-02 10:56:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80ad11ac79e6885c408ea13ccf23a94b053d0eb3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-28 10:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03e12fa6da12082254853410ba510841fb5fa607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:25:55.000000000","message":"Patch Set 1: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c73ec7e53c4b97ba735af3eb6bac5b : FAILURE in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14c99f16e8a94fa4840852350ee6c779 : SUCCESS in 13m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23013b15710c4730b26914c039d83261 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232bc8ec74fe4a2aab76c8ae2fe16cf5 : SUCCESS in 8m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/540ea7c1eb0f45c290c4cdacdd91881b : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ede2a89d2df4b76b5758dfa832e8e84 : SUCCESS in 11m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e224a31698f24fb092117f324907228c : SUCCESS in 45m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d839bbcecb544438b576faf719d5d269 : SUCCESS in 37m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e08bf9d0df41415cbffaf9292348876f : SUCCESS in 43m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/19dd4ca3f4a74151b0018525e766cdd0 : SUCCESS in 40m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9ebeb2d306504dc5898f9925ee9bb7d7 : SUCCESS in 41m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/58872a636c2649078b383ea5e1847c9d : FAILURE in 32m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ce57cb49579247f49a0d95e0999f7161 : FAILURE in 30m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6a2936bf5f1141358b7b68daffc44653 : SUCCESS in 1h 46m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dc9d0b9c9a24493aa0ce8601d4a3074 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"bce1a806cbdde20ee5810d98d1d308f5a4964c08","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-29 05:41:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"299ea0e0f63b08ce1d0a0e1cf08520917ff8ede9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 07:24:49.000000000","message":"Patch Set 2: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c53be6654d402995041ab9f368bcf9 : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efc8952f206c43b994bf50c2bafd8ff5 : FAILURE in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22ddaa1f2994390929ed95623c11c53 : FAILURE in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ee7de653bf446eba4500488fde0831 : SUCCESS in 8m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68012916d1c545c7b6068125e0c7d025 : FAILURE in 8m 28s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bb9cae5063c4db4b7e3ebfad36e0333 : FAILURE in 9m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8cf1e8f292384ac7a67c00459c4293ae : SUCCESS in 54m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64711de3ed744c6ba1da2d82cf6c34e1 : SUCCESS in 45m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77f874ebdb3d4004907aa257168846e4 : SUCCESS in 47m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e0e4c44e94d547c0a8bc452a9360da9e : SUCCESS in 54m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9c4faa0104ad40f086866433e83dd494 : SUCCESS in 44m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9d164b83920f4861b07fa28e5b5d1ab4 : FAILURE in 30m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/24aec3c48a9f4f4a92844208756c4809 : FAILURE in 27m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/31e0bae548894239b3099987bff18e0d : SUCCESS in 1h 37m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55f09cfb845e4a35ab1e37167de541d8 : FAILURE in 9m 04s","accounts_in_message":[],"_revision_number":2},{"id":"346321242f1f63df17ef79eda80b2647ae5d0a21","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-29 07:49:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e2c444ae625372539ca9ccf16ba1f6894167ace","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-29 08:07:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e5a4049d2184b09c3f35500ddfac1eb7ecbe75eb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-29 09:21:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"10bfc6eba133a3be18bba968d10b9b9b96957abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:58:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80c791665804b629bac5d133e022b57 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75feb5b58ab48c2a955f59aefa75be1 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c6f016c2ccb46229785e52828858a9a : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04b9c21dfda48e5abcae39768a68daa : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1c32e6798794c14879a69732ada7ec3 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b8b8836b01f4b9a91fd55a147358d56 : SUCCESS in 8m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8052e2e4a9564f499fc130633eb301d8 : SUCCESS in 33m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a829291ddc4a7295f94e7927db3484 : SUCCESS in 39m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3201b2d73c5443578f6fd8bb8d292f6a : SUCCESS in 30m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f7f4b1bcadbd4b1496088237b6c8e983 : SUCCESS in 38m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/721edf147c1c4fae91eb9276c45dd347 : SUCCESS in 45m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3b6b6605bfb84d9bbb7ffc49c16b0d51 : SUCCESS in 28m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/955787996abe434ab87dc48b74a039ec : SUCCESS in 30m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d75339d13bf44f6ab932fe5977d5e423 : FAILURE in 1h 35m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/174011e50b014fdcbfce3e20ef9c6b10 : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":5},{"id":"9c4e32b2ba748fb9079e4c685fc37cda164e43f3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-01 05:54:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5c14309cc890fa8481677faa8f1edc6434334995","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-01 06:26:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b92118a0355f7c5256c9f6ee1a9c4a5ef21a93c1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-01 07:03:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8dbad43a930f76b1bffbc51dd23a7e87fbb5b5e0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-01 07:06:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"eb0bb85afcce0e256ece88d0ee728e93ac1048d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 08:52:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ccc6c649c954378a01e89583bc12515 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/937a91343557473bbbaadd040a6d2a24 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9b977c5b2b4428a656e030db29bd8c : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3295888d85a643cb85333fe3b14e7197 : SUCCESS in 7m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85df1861e1b046f9a326f605970655e2 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/390146441d9948478709cf0934d6c2d6 : SUCCESS in 9m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e3276f390df14f0e9142d7f2807a6aa1 : SUCCESS in 42m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e63836ba221b4638a5fb0b789b8ee517 : SUCCESS in 45m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d46ada3c22ba4dbcb3370371dcc65a59 : SUCCESS in 43m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/36e3bed98699442f9ace1153f8d9b4f2 : SUCCESS in 44m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2a1f4c22102241c7a61eed30a7b988c5 : SUCCESS in 43m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9489250ce63f4513b254855fa6fcc913 : SUCCESS in 29m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4e7bc92e8d10406aa80eca5a5a4a769d : SUCCESS in 28m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4a35abfdde8d4d3c9f0fc7ac049e95d9 : SUCCESS in 1h 41m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc2e8bb5907041db998248bde8e600f8 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":9},{"id":"dbdc74580f665bdb9144c9683c84399562d0cf2c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-06-01 08:57:23.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"bdf9e334a99e70174379d50b621eb95c21de66cf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-01 09:20:26.000000000","message":"Patch Set 9:\n\nThis is the first patch in the series of refactoring patches that I planned. Please let me know if that\u0027s ok with you. Some things may seem not finishes but that\u0027s on purpose. There\u0027s still a lot of work ahead.\n\nThanks","accounts_in_message":[],"_revision_number":9},{"id":"14a906e543aba56b4c8f4ff33c74d4f9473e226f","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-06-01 13:46:17.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"11d79efca1dad4cafb631bef2a7727e2086beb32","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-06-02 07:02:37.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)\n\nWoow, that\u0027s a huge patch.","accounts_in_message":[],"_revision_number":9},{"id":"2583386ac22ccfd6f182658a7cd7ba4dd47f7ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 07:02:50.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cf31bd3fec724fb7c0edd8a70bd3db3f5a57c133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 08:03:28.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b185ee026094c6e8a0a2dfd44759803 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8888ee33dbf4722aa5f7cd795e9f044 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e5584e063241a98201702bfb315ecc : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af42aefb8724542937689e76835ebb1 : SUCCESS in 10m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3bdba2fe662b43638c3986ebe433c0c0 : SUCCESS in 37m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a3b99a7c124d188ed10d01421373c9 : SUCCESS in 59m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f45107cdb4946ce832b6462380f0c12 : SUCCESS in 39m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b83c241b2f3f4650bd00a1bb8abba272 : SUCCESS in 49m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4f7c70272e0446daae5fbb725c1ee4fa : FAILURE in 30m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a70c5f2c30764394bbd480a50df23922 : SUCCESS in 31m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7c5792168b7d4678a91e1cb43068e448 : SUCCESS in 38m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5abdcf6dffbc4775bf8e571b4a1bc05f : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":9},{"id":"26a61b9151fd6cd36edd8d3cb5d7beeac98d569d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-02 08:33:57.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"15dccc735ce9bb88af88955a8a6b5f6d9058e909","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-02 08:52:03.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c61b1fbc9b7171d3831da41d86b5c99d65ea8563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:04:02.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2aa990433ac4bd39f99ed835fc6c431 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8751625dc72496b94dfa1c52413713f : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c354ea2b17014171abbae60738d90a66 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769a9befe8ea45f1b6710d704453dc5e : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84052fba77a4f9196dabac3cf002fb8 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05c38b7e77f74b64a3e813e225a1845f : SUCCESS in 8m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c455d7151ccd46ac833621616a68bfe7 : SUCCESS in 27m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc618561cfa34241ae99a779eb8cc710 : SUCCESS in 28m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/570780c1917d4faf8f144230c5c85021 : SUCCESS in 28m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/048e0af90d9e4312985a72edb2b29844 : SUCCESS in 27m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d96ca5b19d1a446990ff64e498610fea : SUCCESS in 28m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/37b7d7de0212464aaff4c633a6bdaceb : SUCCESS in 27m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/645bf77405fe4c7e93d2696244a2423c : SUCCESS in 27m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fbbd4ef3edb14f4aab74ba0ca1fb2976 : SUCCESS in 1h 10m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/240007935c9744dab4e3e32cdc9e3bbd : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":9},{"id":"07207b0319bf943cf07f90ba889ab9b5997bf1b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:04:13.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"187fbccbc685dfcfea977581b65a0cc9878103d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:56:54.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f90fb4de62410ebe226e4560f6d621 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d0c6cc20d184487bb392daaa72e6fde : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2193660774a94f08a77d7812e8fa1048 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a85a7ca6a864134936226031b02ba18 : SUCCESS in 6m 36s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6ec861afb10b4a52816404369edeb1d4 : SUCCESS in 41m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0b75cd8ed2947168388a928aaeeee81 : SUCCESS in 42m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bdc7912f33f44108b7ea2a7a73cd1a3 : SUCCESS in 48m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f02583c64aa447a8acc92a2b516eee87 : SUCCESS in 42m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b18bde47775b4fdf8671cbe50c9b02d3 : SUCCESS in 31m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d6b0b61c89f4c0192ac32365a0f4343 : SUCCESS in 30m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fe6331343c1b4b789b69d582dd3a5f39 : SUCCESS in 50m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afa4a67722aa491dab973d62f988d886 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":9},{"id":"12f8c67d2eb4de47441a92c21fcea7598e00a4e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:56:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"4794eeea8017327d6837a3cb790ec7d7d2c2c612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:58:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8c0899b34b41b5a4a39f1f32deb0c6 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"ddf957778533103a29f0887ec02fd70b21ef1e15","revisions":{"462114c77e2b737c4850004db21e61311d0c5e00":{"kind":"REWORK","_number":1,"created":"2020-05-28 10:37:32.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/1"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/462114c77e2b737c4850004db21e61311d0c5e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/462114c77e2b737c4850004db21e61311d0c5e00"}]},"branch":"refs/heads/master"},"b2b6e16382e0aa63d89a557c945530390934ec0f":{"kind":"REWORK","_number":2,"created":"2020-05-29 05:41:48.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/2"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-29 05:40:52.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2b6e16382e0aa63d89a557c945530390934ec0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2b6e16382e0aa63d89a557c945530390934ec0f"}]},"branch":"refs/heads/master"},"4d015ad9680f7711f905719eee1fd1f99dd29b13":{"kind":"REWORK","_number":3,"created":"2020-05-29 07:49:16.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/3"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-29 07:49:01.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d015ad9680f7711f905719eee1fd1f99dd29b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d015ad9680f7711f905719eee1fd1f99dd29b13"}]},"branch":"refs/heads/master"},"47a76a4f95fbe4bfc14e5832c9c5064b1690f3de":{"kind":"REWORK","_number":4,"created":"2020-05-29 08:07:22.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/4"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-29 08:07:08.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47a76a4f95fbe4bfc14e5832c9c5064b1690f3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47a76a4f95fbe4bfc14e5832c9c5064b1690f3de"}]},"branch":"refs/heads/master"},"1bf4a628fd9fcb67b74b001a5b0b4a7ecc33b1d1":{"kind":"REWORK","_number":5,"created":"2020-05-29 09:21:44.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/5"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-29 09:21:30.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bf4a628fd9fcb67b74b001a5b0b4a7ecc33b1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bf4a628fd9fcb67b74b001a5b0b4a7ecc33b1d1"}]},"branch":"refs/heads/master"},"3e18fba5ea90eabaa55a710514c9dfc8ee1a5b6d":{"kind":"REWORK","_number":6,"created":"2020-06-01 05:54:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/6"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-01 05:54:17.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3e18fba5ea90eabaa55a710514c9dfc8ee1a5b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3e18fba5ea90eabaa55a710514c9dfc8ee1a5b6d"}]},"branch":"refs/heads/master"},"b0f15df201df58aca0e57a1f0a9a30785c608f26":{"kind":"REWORK","_number":7,"created":"2020-06-01 06:26:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/7"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-01 06:25:54.000000000","tz":420},"subject":"WIP: refactor policies","message":"WIP: refactor policies\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b0f15df201df58aca0e57a1f0a9a30785c608f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b0f15df201df58aca0e57a1f0a9a30785c608f26"}]},"branch":"refs/heads/master"},"69541a9ccae1c27f0eeb7b4e17d2646a4a5b0b05":{"kind":"REWORK","_number":8,"created":"2020-06-01 07:03:47.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/8"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-01 06:50:57.000000000","tz":420},"subject":"Refactor task policies","message":"Refactor task policies\n\n* The purpose of this patch is to improve encapsulation of task\n  execution state management. We already have the class Task\n  (engine.tasks.Task) that represents an engine task and it is\n  supposed to be responsible for everything related to managing\n  persistent state of the corresponding task execution object.\n  However, we break this encapsulation in many places and various\n  modules manipulate with task execution state directly. This fact\n  leads to what is called \"spagetty code\" because important\n  things are often spread out across the system and it\u0027s hard to\n  maintain. It also leads to lots of duplications. So this patch\n  refactors policies so that they manipulate with a task execution\n  through an instance of Task which hides low level aspects.\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/69541a9ccae1c27f0eeb7b4e17d2646a4a5b0b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/69541a9ccae1c27f0eeb7b4e17d2646a4a5b0b05"}]},"branch":"refs/heads/master"},"ddf957778533103a29f0887ec02fd70b21ef1e15":{"kind":"REWORK","_number":9,"created":"2020-06-01 07:06:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/91/731491/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/731491/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/731491/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/731491/9"}}},"commit":{"parents":[{"commit":"c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4c11c9b85eb5a3fbe76cab6a65cdd0aa32c3e63"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-05-28 10:37:12.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-01 07:05:49.000000000","tz":420},"subject":"Refactor task policies","message":"Refactor task policies\n\n* The purpose of this patch is to improve encapsulation of task\n  execution state management. We already have the class Task\n  (engine.tasks.Task) that represents an engine task and it is\n  supposed to be responsible for everything related to managing\n  persistent state of the corresponding task execution object.\n  However, we break this encapsulation in many places and various\n  modules manipulate with task execution state directly. This fact\n  leads to what is called \"spagetty code\" because important\n  things are often spread out across the system and it\u0027s hard to\n  maintain. It also leads to lots of duplications. So this patch\n  refactors policies so that they manipulate with a task execution\n  through an instance of Task which hides low level aspects.\n\nChange-Id: Ie728bf950c4244db3fec0f3dadd5e195ad42081d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf957778533103a29f0887ec02fd70b21ef1e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf957778533103a29f0887ec02fd70b21ef1e15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
