)]}'
{"id":"openstack%2Fmistral~732860","triplet_id":"openstack%2Fmistral~master~Icaf1922963f4bc2fd2c4138aa3fe15cce5d3f1f7","project":"openstack/mistral","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"Icaf1922963f4bc2fd2c4138aa3fe15cce5d3f1f7","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 18:23:47.000000000","updated":"2020-06-03 05:01:39.000000000","submitted":"2020-06-03 05:00:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732860-1591160427535-08d79f7f","meta_rev_id":"0c8fc254ea3f84b3ff90036c4b602e25243bfe6b","_number":732860,"virtual_id_number":732860,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 05:00:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-06-03 04:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-06-03 04:29:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 04:29:37.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-06-03 05:00:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0387ebf2b42fbcc1d98fbebc83f55afc27e304b2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 18:23:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8523c1eccc7a86a08db112009f1da22c3ed72878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:58:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431ef68f5ebf461e91d64085c87fd38f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95652c948b2741b4ba322374ee57c84e : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c521ff01dd548ebaf38a8859de97e0d : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70964e191fbb4282883e917eb7987c61 : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64abb0645a2a487ea4fb24ab55d45876 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af3e6c887b744f0ca8217910327a7d60 : SUCCESS in 10m 32s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5c6e74a3d6e846b6b99c1dcc99c0a695 : SUCCESS in 42m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee10f445da29464cb084375fbe4362e0 : SUCCESS in 39m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3ecb25920864756a8e57213481f48b2 : SUCCESS in 40m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/243bf5ba9d184a58a740d1677a0e577d : SUCCESS in 45m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/21ed74c6ef6e4dc680db133b9f75752e : SUCCESS in 43m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a1884f573a7d40789ee44961c6c5af84 : SUCCESS in 33m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c9ff34675dfe46f3977e09f585ab7c18 : SUCCESS in 30m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/cb10d76d66c6472db6cdd5335be5731a : SUCCESS in 1h 33m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7e9574b15a64bb19003ef87ca71026f : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7f6bc4d9d8e4671247421939afcfaebb0716086d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-03 04:09:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb79ed0a96a619d777513a281f75a197053408ab","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-03 04:29:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bccb3e3be48d40708710c3bfc812d321cf166ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 04:29:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a84157c0d66b157c1de3c4289db32f57a636cee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 05:00:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ef11ddc3483e91b79280036c27ef2feb794b3960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 05:00:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20bd580730c343e396ab02806cca46a0 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/791daa4a01a54540a4457750c7b47a6d : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85bb6b10a5ab4395b1135dcea2cabdeb : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218d46fdcbe04811a921e80ec57ce566 : SUCCESS in 5m 51s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dd3d2e16112c4601a345adf46b0cd35f : SUCCESS in 29m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad173c33436f4b29983eb4770a4a8d9c : SUCCESS in 27m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acd514f539714e0cbab055a6ace310ec : SUCCESS in 28m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1a2da1ac7e8a441bbd3e02ec36148e71 : SUCCESS in 29m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b1c3add132b544e694d3e677e9510c00 : SUCCESS in 27m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d148b883e0334cc3bc5971a32d24c247 : SUCCESS in 26m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/455f10d0fd9041b6bc98a8c7b0021a88 : SUCCESS in 28m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5fd6773407c427d9ad60dd51bdf764c : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1968efe996e467832ece3f47d6af84558e281fef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 05:01:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edb7c7d0af44d909755c686f364f5b6 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"986277d60793b26e3bb629ba458b5c75e119fbc5","revisions":{"986277d60793b26e3bb629ba458b5c75e119fbc5":{"kind":"REWORK","_number":1,"created":"2020-06-02 18:23:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/60/732860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/732860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/732860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/732860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/732860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/732860/1"}}},"commit":{"parents":[{"commit":"ddf957778533103a29f0887ec02fd70b21ef1e15","subject":"Refactor task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf957778533103a29f0887ec02fd70b21ef1e15"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:23:39.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:23:39.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: Icaf1922963f4bc2fd2c4138aa3fe15cce5d3f1f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/986277d60793b26e3bb629ba458b5c75e119fbc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/986277d60793b26e3bb629ba458b5c75e119fbc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
