)]}'
{"id":"openstack%2Fmistral~733092","triplet_id":"openstack%2Fmistral~master~I9c0647235e1439049d3e7db13f19bef542f10508","project":"openstack/mistral","branch":"master","topic":"refactoring","hashtags":[],"change_id":"I9c0647235e1439049d3e7db13f19bef542f10508","subject":"Refactor task notifications","status":"MERGED","created":"2020-06-03 10:33:00.000000000","updated":"2020-06-04 13:59:51.000000000","submitted":"2020-06-04 13:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733092-1591279072788-2e1badb8","meta_rev_id":"38e4737d5a8b1d20af56bda03ec0508740425486","_number":733092,"virtual_id_number":733092,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-06-04 10:45:51.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 13:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-06-04 09:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 09:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-06-04 09:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-06-03 12:01:37.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-06-03 12:01:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-06-03 12:01:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 12:01:32.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-06-03 12:01:32.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-06-03 12:01:37.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-06-04 09:25:26.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-06-04 09:33:55.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-06-04 13:57:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a7696085ff360f6483359a52f805629a5a4c941","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-03 10:33:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"753325e48fc3bdf5479933b4d4ad22bd8123d7fd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-03 10:52:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f42860347d62cbc0e547296bb76455f6551e349b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:54:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab54ba10d6f45679def5a2590bf17b9 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3db8887b002149d683341a4eeb0cd85c : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8a53dcdede8417bb8ae30a4655cd8bb : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd08a5b2a794223b26a21f77f520f8c : SUCCESS in 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/329400c543df4926a9e6d82551753264 : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/637fcc4045044d1a9841e8ec72b2ddd8 : SUCCESS in 10m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0545d1ae67744a1db5057b52a1602c73 : SUCCESS in 44m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74dba49d1c4c49988ec2a1b0b08a827a : SUCCESS in 45m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/453517acdc5b4aea89ed3c86507e47c5 : SUCCESS in 50m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1d73d417f8b14c1baef82914a6835d35 : SUCCESS in 44m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9c2fffee462042d58f4b0709575da110 : SUCCESS in 43m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6c402766c670432686d639dcbdfcb5df : SUCCESS in 30m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ed1236ebb56f4dfe806d00a520102319 : SUCCESS in 28m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5267228f757349d98c3b4b37af4eb699 : SUCCESS in 1h 56m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f5ed6195edb43a18ffd0771a6691753 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":2},{"id":"06ac0aa16d49252fe877bab4afcbea78d41ed5b1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-06-04 09:25:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1eb78996d3f9a86aaded15737298792e5c80576c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-06-04 09:33:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47cbe166283471a2f18adab93901ad7f668bc569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:34:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"963079dbb0b45dd9a90cfe1e4fcd1fa35f76b6da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:33:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28652f5b26dd48aab0f80b2c9cd40034 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd59f77cf5af44e1bc5cca2019023c8b : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44f8b6365ee64adbbdb3f638af381d26 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f7a3014167b448a92417bed2aad51a9 : SUCCESS in 9m 38s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b1f11d8d4a4540c9ba5b2ca81d7d0959 : SUCCESS in 44m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f762b197ecf24d799418085609662283 : SUCCESS in 49m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e9e59ddc4743e09372acb5a6ea7435 : SUCCESS in 52m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/244df6f52a754c6881313a9507d99624 : SUCCESS in 49m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9690603a423f4d97b775ad58e2672428 : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3c27a5ea9b59402ba22f03a143f901d8 : SUCCESS in 27m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f7277e1460874853acf8fbcf448f9450 : FAILURE in 31m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c28128e0f8104b0e97a458bc2f39a4b8 : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c4dfcaad4f2327245864f15016949218d5b5f19e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-04 10:45:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2016ebfd7425b449e425277bca1e2b51820b4892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:42:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f048b0899d844a519f1c29906d89bc97 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/577796898b0d43cf9c16e42832e6ecc7 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7235d45054d40419f1b3e2e4666ff4a : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba21d4087d234430a4077573ffd7c627 : SUCCESS in 10m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a82381323efb4c8f857a51ba7d0419fa : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0d2a48cf3d341aebe99d625a66d13f4 : SUCCESS in 10m 53s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7ddfdefc897c45ee9357108c6e83ac90 : SUCCESS in 52m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb9d7a91548d4caa9268e208ebfaddb8 : SUCCESS in 50m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a521c26e94554530a1c16a593e0c3ddf : SUCCESS in 52m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5739e303848949b9b5521de7b87143f1 : SUCCESS in 41m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d945a6f1ea5b4f768fb54aa3661e663f : SUCCESS in 50m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/73bff03cc24a4a1eae981834c0ad2708 : SUCCESS in 32m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0ccc3ba8600a4224a0b18a4a36f0dcd7 : SUCCESS in 33m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0f1ec589c2f246c18a49dc749bd77bfd : SUCCESS in 1h 49m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5388df7c21e1476886668bbcc688004d : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f199601400ce7ea370f2c4a4d094c6100a4c5901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:42:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70ba7e96081b21246f524d730bdaffc15c524297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:57:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ebe5fbd44041678b14dfbdb04b315e : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aea20ffc8c440319b085a834338f90f : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8044215e91354252b7a38d575479069c : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd52172660b74502a2e0bd79f0cb77eb : SUCCESS in 8m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f0c57a0d36134d129b0194b483f1fd34 : SUCCESS in 46m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/397c83909335424e846a41114e3da78e : SUCCESS in 1h 13m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07b04153719f439c9d853bbd50616678 : SUCCESS in 1h 11m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d6905fe3127948178d152f3c925553ea : SUCCESS in 52m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2b3490464ca444e6b76622f8ce7227e1 : SUCCESS in 30m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/97e66975afb84f0fb2f407debf73cf6b : SUCCESS in 30m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f3685339297644f290285ba4718ae2a9 : SUCCESS in 33m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/850c829b813b4447923f36d1cd791bb1 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d49db1ff015f0a001240e785a0a93a94131cb81a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:57:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8aa60bb66caa022e2aedfc0fb2ea61d7115543b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:59:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a162fc342a4c2aa078e04ef47c0ab0 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b55dbdea68561d65ed3ef52d3ccb7bdbc1a283f9","revisions":{"790ecaf213f6b1c700bb670b8e95711f80e88dab":{"kind":"REWORK","_number":1,"created":"2020-06-03 10:33:00.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/92/733092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/92/733092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/92/733092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/92/733092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/92/733092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/92/733092/1"}}},"commit":{"parents":[{"commit":"a620dabb78fb52e83c2544f08ce0546f89603abf","subject":"Simplify setting task \"started_at\" and \"finished_at\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a620dabb78fb52e83c2544f08ce0546f89603abf"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-03 08:37:56.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-03 10:31:40.000000000","tz":420},"subject":"Refactor task notifications","message":"Refactor task notifications\n\n* All calls to a notifier within the Task class have now been\n  moved into the method set_state() so that the relation between\n  a state change and a notification is now straightforward and the\n  notification calls don\u0027t have to be spread out across different\n  modules.\n\nChange-Id: I9c0647235e1439049d3e7db13f19bef542f10508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/790ecaf213f6b1c700bb670b8e95711f80e88dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/790ecaf213f6b1c700bb670b8e95711f80e88dab"}]},"branch":"refs/heads/master"},"b55dbdea68561d65ed3ef52d3ccb7bdbc1a283f9":{"kind":"REWORK","_number":2,"created":"2020-06-03 10:52:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/92/733092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/92/733092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/92/733092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/92/733092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/92/733092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/92/733092/2"}}},"commit":{"parents":[{"commit":"a620dabb78fb52e83c2544f08ce0546f89603abf","subject":"Simplify setting task \"started_at\" and \"finished_at\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a620dabb78fb52e83c2544f08ce0546f89603abf"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-03 08:37:56.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-03 10:51:41.000000000","tz":420},"subject":"Refactor task notifications","message":"Refactor task notifications\n\n* All calls to a notifier within the Task class have now been\n  moved into the method set_state() so that the relation between\n  a state change and a notification is now straightforward and the\n  notification calls don\u0027t have to be spread out across different\n  modules.\n\nChange-Id: I9c0647235e1439049d3e7db13f19bef542f10508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b55dbdea68561d65ed3ef52d3ccb7bdbc1a283f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b55dbdea68561d65ed3ef52d3ccb7bdbc1a283f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
