)]}'
{"id":"openstack%2Fmistral~733470","triplet_id":"openstack%2Fmistral~master~Ib8f61481a606fe4fc9f37112ef625b8e3c6d5cd3","project":"openstack/mistral","branch":"master","topic":"refactoring","hashtags":[],"change_id":"Ib8f61481a606fe4fc9f37112ef625b8e3c6d5cd3","subject":"Remove one more self.notify() call from class Task","status":"MERGED","created":"2020-06-04 07:45:57.000000000","updated":"2020-06-05 07:10:07.000000000","submitted":"2020-06-05 07:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733470-1591340922743-63743c72","meta_rev_id":"b5068e20e54e3331c7fb360b96d321e88e47935d","_number":733470,"virtual_id_number":733470,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-06-05 04:12:36.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 07:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-06-04 09:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 09:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-06-04 09:13:20.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-06-04 09:13:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-06-04 09:13:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 09:20:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 09:13:13.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-06-04 09:13:13.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-06-04 09:13:20.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-06-04 09:19:47.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-06-04 09:20:39.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-06-05 07:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ece99fec2091a9e5c0e73ee0b3c3101b1ea4d92","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-04 07:45:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e2816f1c50ea6b58af629b31e9d2b9339075280","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-04 08:19:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5cbef9660702f3658fa3379fc8f9bcd55b9b1232","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-04 08:31:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11fc295ab85f6c79863ff6d09b4b61f2d8a30477","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-06-04 09:19:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2459046d8b2ee61ad88c747ca04c5493dcc18dd3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-06-04 09:20:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c01e5ff1559b53c01968fc8bb5083de745d3089b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:09:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee68ee001a1467cbbac346515b311e9 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2c7d11c48ed4d65856b6f6baed796c9 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/045b0e0b97b042cd8e99a4448ba2e070 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94497d4b68da4708990f36805a9928b2 : SUCCESS in 11m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b76fe5db0b0946eabfe50240924ded0b : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b779904710a418991af7fd66a034c99 : SUCCESS in 11m 21s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/26b2cd778a094c97a1066d1c4b9be82a : SUCCESS in 32m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f02019ff6f134698827c43eaabb09c4c : SUCCESS in 28m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/452e98a03d954ef184fdda63a706cb43 : SUCCESS in 32m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a9322dac20734ac6b72e55e7f8fb3d7d : SUCCESS in 29m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4039584ff1ab474aa806b9d95580d029 : SUCCESS in 31m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ba1d5362484f44ee9f112359b7a870a3 : SUCCESS in 28m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1d005c5930dc4e1c8fb4b341a9ca415d : SUCCESS in 29m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f9bb784a917844ccb13aa2b9bc580609 : SUCCESS in 1h 36m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9219cd9afdd443cb7798f73b216f36b : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":3},{"id":"e1644a3f77a764d4e68fbc78d3ea27ad8c3f0f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:42:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"682eb2019d50a29bedde905c83e5633209848c5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:57:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5117e31f71422c8cdc072cb5bca3b1 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/190a35064f3e4fc08cddb79c89f63860 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2398cb8eea4846259cd34fcd55ec1a2c : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06812d96eb4d413ab84593fd3388d72b : SUCCESS in 7m 37s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/170a27305b7f49848cc8cc0271a717b9 : SUCCESS in 43m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c75b07211c6f474f82d53cab1de916bb : SUCCESS in 55m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2f700b9fae74ec685646e3eb71d28dd : SUCCESS in 47m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9fab6c5726e34dac92c1fe3a9240a2b3 : SUCCESS in 51m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ac5f4a9466a046bea9b6b0a8d98258da : FAILURE in 33m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/23f5aa175a624e7f9785c18dd084d5dc : SUCCESS in 27m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d869bf251b2e4a9ebf9d4ae945f3c160 : SUCCESS in 49m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b16219fdec9c4e95b831bb69b1756bf1 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b1cf8eeb6b54ddfe2de309b24cebd354c2fd43c1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-05 04:12:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec26cff07b3cc6339dbbc9d450a01836b88f0535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:13:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729a419708f74582ae01237df1f7bd94 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9944fd45d91423086dbe723f4dd7609 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d7ca4918b6a496fadd6428af60bab63 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0ee9741ae14c568c533ff45eeb0946 : SUCCESS in 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51849cc855694ca8bf946258f2625f28 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/837b952ba00449f887ad1d5bdbb9b2d7 : SUCCESS in 10m 42s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/caafc462cb8e4046b400fe3d462e9119 : SUCCESS in 37m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bd28d1b2a3f4fb8a9433b5d3d9a4cf0 : SUCCESS in 45m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eeaa08b6b6649e1b38ba6b6b3de731a : SUCCESS in 40m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c6966fdb42c4b15a593c0fd34988499 : SUCCESS in 36m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/17d3a6603ac848c2b550c14c21fc5cb2 : SUCCESS in 40m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ec3703f45c7c4f4e9615f38b0d84fc36 : SUCCESS in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8cc6c0a960c84b7c977d95ef7a2bae58 : SUCCESS in 30m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/02eaa13f23fd4f0eafdcbd4f6a0df248 : SUCCESS in 1h 55m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73cbd2100d4b42a18091d96bf7a816be : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":3},{"id":"7ad56577a6f7362a2d4a564ca39d3ede6c71dc45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:14:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1bb38580f824698359d5d89d9a531d4c9f4cd3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 07:08:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f813a076b86a4f12adc063891b9af82c : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fa4e120e5ef4715b9c5bcc41f30b352 : SUCCESS in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7773c18bb9a249abbc42408fb2e81904 : SUCCESS in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ae543e04b54b28ba455c082e153438 : SUCCESS in 12m 25s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f5703e9c3d5b4301b26f28302e997193 : SUCCESS in 44m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/434c0aa4162a4372b017fbacc43bc61c : SUCCESS in 50m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5fac427ab4c4513aae2462996ac0f6d : SUCCESS in 41m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/aa86d15464a346438cd8df158bc5f46f : SUCCESS in 42m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/80a4a4480a7b43c1a0230fb85ef845a1 : SUCCESS in 32m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/835e8f177ab84c0dbbd5d9f308c765e2 : SUCCESS in 31m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dd4a994e28b1452280355537f7263ce1 : SUCCESS in 47m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2409b65a395347028ad6c72904d0bdc2 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":3},{"id":"847f0efdba47fa6fe18d8dd7de1157228b77a3e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 07:08:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dc01fd9df6fa0e814502c9d98379a6e93e061b9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 07:10:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82ba90fd6cc439d9041dcd98d47d05d : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b7a2bba01ccfb562531eb50ce86032aaedbf330","revisions":{"1efab2b8f4da4b888fc594ad21699f91d2df0471":{"kind":"REWORK","_number":1,"created":"2020-06-04 07:45:57.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/70/733470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/733470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/733470/1"}}},"commit":{"parents":[{"commit":"90d1f1ba8e479d8ead8b7b458268be52849d0a16","subject":"Refactor workflow notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90d1f1ba8e479d8ead8b7b458268be52849d0a16"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 07:37:03.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 07:43:25.000000000","tz":420},"subject":"Remove one more self.notify() call from class Task","message":"Remove one more self.notify() call from class Task\n\n* Getting rid of another self.notify() call in the Task class\n  that\u0027s not inside of the set_state() method. That also gave\n  an opportunity to not manage started_at timestamp out of\n  the method set_state().\n\nChange-Id: Ib8f61481a606fe4fc9f37112ef625b8e3c6d5cd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1efab2b8f4da4b888fc594ad21699f91d2df0471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1efab2b8f4da4b888fc594ad21699f91d2df0471"}]},"branch":"refs/heads/master"},"3d1390a77b03af8e50cded2e0c41efc4ee13ca2a":{"kind":"REWORK","_number":2,"created":"2020-06-04 08:19:29.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/70/733470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/733470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/733470/2"}}},"commit":{"parents":[{"commit":"90d1f1ba8e479d8ead8b7b458268be52849d0a16","subject":"Refactor workflow notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90d1f1ba8e479d8ead8b7b458268be52849d0a16"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 07:37:03.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 08:18:37.000000000","tz":420},"subject":"Remove one more self.notify() call from class Task","message":"Remove one more self.notify() call from class Task\n\n* Getting rid of another self.notify() call in the Task class\n  that\u0027s not inside of the set_state() method. That also gave\n  an opportunity to not manage started_at timestamp out of\n  the method set_state().\n\nChange-Id: Ib8f61481a606fe4fc9f37112ef625b8e3c6d5cd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d1390a77b03af8e50cded2e0c41efc4ee13ca2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d1390a77b03af8e50cded2e0c41efc4ee13ca2a"}]},"branch":"refs/heads/master"},"2b7a2bba01ccfb562531eb50ce86032aaedbf330":{"kind":"REWORK","_number":3,"created":"2020-06-04 08:31:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/70/733470/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/733470/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/733470/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/733470/3"}}},"commit":{"parents":[{"commit":"90d1f1ba8e479d8ead8b7b458268be52849d0a16","subject":"Refactor workflow notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90d1f1ba8e479d8ead8b7b458268be52849d0a16"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 07:37:03.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 08:31:41.000000000","tz":420},"subject":"Remove one more self.notify() call from class Task","message":"Remove one more self.notify() call from class Task\n\n* Getting rid of another self.notify() call in the Task class\n  that\u0027s not inside of the set_state() method. That also gave\n  an opportunity to not manage started_at timestamp out of\n  the method set_state().\n\nChange-Id: Ib8f61481a606fe4fc9f37112ef625b8e3c6d5cd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b7a2bba01ccfb562531eb50ce86032aaedbf330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b7a2bba01ccfb562531eb50ce86032aaedbf330"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
