)]}'
{"id":"openstack%2Fmistral~733563","triplet_id":"openstack%2Fmistral~master~I8c7c6896f01894ac76cf9365abfdce17e7df7662","project":"openstack/mistral","branch":"master","topic":"improve_actions","hashtags":[],"change_id":"I8c7c6896f01894ac76cf9365abfdce17e7df7662","subject":"Moving task type constants into Mistral task specifications module","status":"MERGED","created":"2020-06-04 11:23:49.000000000","updated":"2020-06-08 13:18:51.000000000","submitted":"2020-06-08 13:17:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733563-1591622231531-59896c92","meta_rev_id":"0d8e8e260d544337265688568f861ed18caca79e","_number":733563,"virtual_id_number":733563,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 13:17:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-06-08 12:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 08:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-06-08 12:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-06-08 07:04:13.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-06-08 07:04:07.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-06-08 07:04:07.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 07:04:07.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-06-08 07:04:07.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-06-08 07:04:13.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-06-08 08:34:16.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-06-08 12:24:40.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-06-08 13:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7537cbd6a4af8f263149ccb615f7831b2d18af1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-04 11:23:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5019578fd053364706939994a67c5b175c6ff1b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:13:09.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc45c00bbac4bb1a09fe345f0c5a13d : FAILURE in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c4ce51886a0443ba62d8910f89e52f4 : FAILURE in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db171a438cba442aab96d47fc2c624f8 : FAILURE in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1be63e2ce104427afe9c7414c2cfd7a : SUCCESS in 9m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7144772b25a442edbe1b94af21a18049 : FAILURE in 9m 29s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbf11df7e47a4d2f80d27560f44693e2 : FAILURE in 10m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/14d9468293aa4b68a62c81878001d405 : SUCCESS in 50m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea6808baa6934770be56109e235bf89b : SUCCESS in 45m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f4d1caa45f542ac844ee0cb717c1d06 : SUCCESS in 47m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3a55ea4923224d8a9725029774846d88 : SUCCESS in 43m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1e65f84ce4f742d5a2503320c3a18673 : SUCCESS in 52m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0f6db08ee7e94068ba923906a2262102 : FAILURE in 29m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4db85ce126884ed390339d29f8508688 : FAILURE in 29m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0d4eacf961714a44912b984ada8fe2af : SUCCESS in 1h 42m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db1f85f8eef048ecbe6f0638e62dda23 : FAILURE in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3918387ecdc11424fde71b989d8bb607c93e988d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-05 04:13:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a240875f6518277cbdbef3e2076be906e623eb5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:02:02.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42b80eae2774dc183e3e42246db5fc5 : FAILURE in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3872968ab2e24552990055ba3d7b82b0 : FAILURE in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2eeeeca35d449de81ddbda6eb19b22f : FAILURE in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec50d64c1e79408b8e5476b10d369208 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92b77d6eb6a94f20b41e4fa9fb34065b : FAILURE in 9m 50s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4102502ed61843a6b594f0ef2ca01170 : FAILURE in 10m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3269d873c98a4ed78c098e7b5bb45c30 : SUCCESS in 37m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6263f840eb494ac4983886c569ceef1e : SUCCESS in 46m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee705386c0914bf38e536b2d493a9eac : SUCCESS in 44m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0a940af430b1452e82e250af60e89b30 : SUCCESS in 41m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/016f0830a0f3450db611e1cf8e8a6a61 : SUCCESS in 38m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ac98b2569ef94661b76267482e33e9c1 : FAILURE in 29m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c6b8e21fd3cc46b18563b4a28238354c : FAILURE in 30m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4dcdc61ace624dc8850eec8f599b8605 : SUCCESS in 1h 43m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c50278c6504f50b39bbb9281363eee : FAILURE in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c494360019aab99ece145928a52e9d26ab03c352","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-05 06:12:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fde407d2b4da837f802a01eb97c664926d8f9bc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-05 07:58:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"15cfa3e53ff2066cfb71525d10e47cec73eae3dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 09:43:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc0c4ef7f194776b9aa16d17ca8a66c : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33e3a8f77b434e948f8305cd9cfa2eb4 : FAILURE in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc605b37f2d44b68866da8c434b4535 : FAILURE in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71b222aecd643e09a12f7ea964aa811 : SUCCESS in 9m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95f43aeb9a094754be4c28af8b04f0bc : FAILURE in 8m 42s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a59ffef87684a48861844497fbb070c : FAILURE in 11m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6e672deaaa794aa0a885db30d1a7cf70 : SUCCESS in 46m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9af599f5987d43c4ac381b06a9ac36c8 : SUCCESS in 1h 12m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ae7130fb9a24ff1a9c3790f5e18bf38 : SUCCESS in 1h 09m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4d116f2174b04aa8b56edcbfd88fc474 : SUCCESS in 1h 14m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dd64042d800f4f278425bf44dc681e3e : SUCCESS in 1h 10m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/38d6cd864c914c53849f44c56d259707 : FAILURE in 31m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8868a37f4781412b9d2319a55644980b : FAILURE in 32m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/efad238f3d25458dba41dadc1f248e9b : SUCCESS in 1h 43m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10214a8c4f0744bd93e5242df4512f22 : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":3},{"id":"36211cd73c012610961b52c33e31fd80878ac7b1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-08 06:44:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d5c155a94b1bbe361991976bf23a43adcfd49fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 08:33:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39808c588b834b529ef76b792d861a8a : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd6ba0f813e4afeb69a715e08664faa : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b58fb0032c7946a3b58290d73afb9a5b : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47872890eea04ec1a5a070b3fc470f1e : SUCCESS in 11m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cffaf325e2b34d2aa2f1451f6c2c7e72 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0a441b7f2a340d588fa74867c2e9358 : SUCCESS in 11m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6641c1b53f9946aba5f3452f6fc16a4a : SUCCESS in 47m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c630f305d61c4281a70e8b6ec36c9ebf : SUCCESS in 42m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25b6ff7c86ca4e5199deede6fd570696 : SUCCESS in 45m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c69607e26b6e43e69f19e9fd7370c347 : SUCCESS in 43m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc6e936b733b4f77ab31a390fc4195ad : SUCCESS in 47m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/df1f46ec2c584d66940a764a5067da6c : SUCCESS in 31m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f3dc41fbf79f4fdd85f4b3196f360367 : SUCCESS in 29m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6590e8081bd94efb9047a4bbba673e07 : SUCCESS in 1h 41m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f99d8bfcda554422b95d60517646fe35 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":4},{"id":"db0e63035a1b2529d4ef7531146f1466fe8cf5c1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-06-08 08:34:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bd27677285916c4f588521f48ca51b19906319b1","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-06-08 12:24:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"da8625b79c7bd1fcc5e07046e2f95b57bad84399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:24:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"702a46e97d9384ca4cdd8aee422c92e91eff2218","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 13:17:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"14cf6234449a86d87b6905ac4b9e3ffcc65b92ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 13:17:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1e689826e74bd4936ee43ab1323029 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7636690eebc444ca9815891e4099320 : SUCCESS in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b177144698459da53bc59360c8ead8 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d16018477f9419d83be84b07465caff : SUCCESS in 8m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/433ef7c7f38b46d78b80e0f4cdbfa436 : SUCCESS in 45m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c5c0972aa1e419992805546747a32dc : SUCCESS in 45m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2c7167af78f4837b6157ccf00c3f33c : SUCCESS in 47m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe94f77ca5be49ac8303f2f8f5dff21b : SUCCESS in 45m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f04bfb31a0114a90abd30ea0513cd24f : SUCCESS in 30m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cc5d7083a24d4c0c81be361f72ad7677 : SUCCESS in 29m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/23f4df6248fd4dcfa19c0f536235b3b5 : SUCCESS in 43m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14b971c7e1f44d8e861ce7acbf25cc13 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":4},{"id":"04fe373b23d21deedc95681066a2370bfdc4a5c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 13:18:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cde4aacd1e5457bbfa3adad8757d50a : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2ad7bb829b3ae713a21d0d006ef9d15085fe85e7","revisions":{"f46b539d8bb06862e3d546430df02b261c2a9898":{"kind":"REWORK","_number":1,"created":"2020-06-04 11:23:49.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/63/733563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/733563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/733563/1"}}},"commit":{"parents":[{"commit":"ddf957778533103a29f0887ec02fd70b21ef1e15","subject":"Refactor task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf957778533103a29f0887ec02fd70b21ef1e15"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 11:19:52.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 11:19:52.000000000","tz":420},"subject":"Moving task type constants into Mistral task specifications module","message":"Moving task type constants into Mistral task specifications module\n\n* There were references to Mistral specific constants mistakenly\n  placed into mistral-lib. This patch adds these constants into\n  the module where task language specifications are declared and\n  changes the corresponding references.\n\nChange-Id: I8c7c6896f01894ac76cf9365abfdce17e7df7662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f46b539d8bb06862e3d546430df02b261c2a9898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f46b539d8bb06862e3d546430df02b261c2a9898"}]},"branch":"refs/heads/master"},"3ac426e3835aa3d9a2a5a78854ede1cb12cd0358":{"kind":"REWORK","_number":2,"created":"2020-06-05 06:12:42.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/63/733563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/733563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/733563/2"}}},"commit":{"parents":[{"commit":"ddf957778533103a29f0887ec02fd70b21ef1e15","subject":"Refactor task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf957778533103a29f0887ec02fd70b21ef1e15"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 11:19:52.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-05 06:12:13.000000000","tz":420},"subject":"Moving task type constants into Mistral task specifications module","message":"Moving task type constants into Mistral task specifications module\n\n* There were references to Mistral specific constants mistakenly\n  placed into mistral-lib. This patch adds these constants into\n  the module where task language specifications are declared and\n  changes the corresponding references.\n\nChange-Id: I8c7c6896f01894ac76cf9365abfdce17e7df7662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3ac426e3835aa3d9a2a5a78854ede1cb12cd0358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3ac426e3835aa3d9a2a5a78854ede1cb12cd0358"}]},"branch":"refs/heads/master"},"be1dd015227380c51ccfe481d15abe9ce964e64b":{"kind":"REWORK","_number":3,"created":"2020-06-05 07:58:38.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/63/733563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/733563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/733563/3"}}},"commit":{"parents":[{"commit":"8953999bc5e007140989d7c8f843d75c8031dba7","subject":"Merge \"Remove one more self.notify() call from class Task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8953999bc5e007140989d7c8f843d75c8031dba7"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 11:19:52.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-05 07:58:20.000000000","tz":420},"subject":"Moving task type constants into Mistral task specifications module","message":"Moving task type constants into Mistral task specifications module\n\n* There were references to Mistral specific constants mistakenly\n  placed into mistral-lib. This patch adds these constants into\n  the module where task language specifications are declared and\n  changes the corresponding references.\n\nChange-Id: I8c7c6896f01894ac76cf9365abfdce17e7df7662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be1dd015227380c51ccfe481d15abe9ce964e64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be1dd015227380c51ccfe481d15abe9ce964e64b"}]},"branch":"refs/heads/master"},"2ad7bb829b3ae713a21d0d006ef9d15085fe85e7":{"kind":"REWORK","_number":4,"created":"2020-06-08 06:44:28.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/63/733563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/733563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/733563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/733563/4"}}},"commit":{"parents":[{"commit":"8953999bc5e007140989d7c8f843d75c8031dba7","subject":"Merge \"Remove one more self.notify() call from class Task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8953999bc5e007140989d7c8f843d75c8031dba7"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-04 11:19:52.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-08 06:40:45.000000000","tz":420},"subject":"Moving task type constants into Mistral task specifications module","message":"Moving task type constants into Mistral task specifications module\n\n* There were references to Mistral specific constants mistakenly\n  placed into mistral-lib. This patch adds these constants into\n  the module where task language specifications are declared and\n  changes the corresponding references.\n\nChange-Id: I8c7c6896f01894ac76cf9365abfdce17e7df7662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ad7bb829b3ae713a21d0d006ef9d15085fe85e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ad7bb829b3ae713a21d0d006ef9d15085fe85e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
