)]}'
{"id":"openstack%2Fmistral~734830","triplet_id":"openstack%2Fmistral~stable%2Fqueens~I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","project":"openstack/mistral","branch":"stable/queens","topic":"bug/1850843-stable/queens","hashtags":[],"change_id":"I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","subject":"Mask sensitive data when logging action results","status":"ABANDONED","created":"2020-06-10 13:14:07.000000000","updated":"2021-08-18 08:54:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e936700694c7344fb40d327f6048ceede264bd78","_number":734830,"virtual_id_number":734830,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-29 11:47:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-09 13:22:34.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-07-29 11:47:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-07-29 11:47:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 13:14:07.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-06-10 13:39:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-06-10 13:39:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-06-17 07:58:24.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-07-02 19:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 12:32:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-07 05:31:30.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"6cc154a62a5f91d79cd5e16f3801d53364cf924b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-06-10 13:14:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"114342dc96dcfe297b40c38021fa4bccbea44298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:33:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734ef3f884ff4128aa14896605a4b0c9 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bb29ecc3113441cae1278a346aa1ce8 : RETRY_LIMIT in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d983c444c5e4a9bbcee270bdce052c4 : RETRY_LIMIT in 3m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/12105a36edaa43009ec3e67c11b108ea : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2267dac71d24ac78ebafd05798bb54c : RETRY_LIMIT in 2m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7114004ca38040e9918f25910bee01ae : FAILURE in 38m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/270cc9967e5e4079ab49ca4ec974a995 : FAILURE in 29m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9ffbf13d69724242af0aedbe7e90575e : FAILURE in 29m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aaaafab15d7645ef802c34650580e05c : RETRY_LIMIT in 3m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bb68fdca12d645e6bb57afac519b7d07 : RETRY_LIMIT in 2m 56s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/552c20f0fdf8405e925ab62e934d5106 : SUCCESS in 2h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbbe432de24f0d5d0319a90a234bbdb8a6c3cd84","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-06-11 07:46:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"01a8ce36d9fad8e2b745d7bfd546cc9cba756694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:39:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba29e4903124b68a2a5a48d6e41791f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2374bd3e944043dc95c82a004f46fc4d : RETRY_LIMIT in 3m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a13cab779ea64338826dd3f873798ea0 : RETRY_LIMIT in 5m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4eab7e1f3c4b4f1886c4cc1651a0bc21 : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d096300fb96421cb11a70a527a81212 : RETRY_LIMIT in 3m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0f384ff935d04b35bab0b815011cc8a1 : FAILURE in 38m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3172120d28674ca19f48b8c18e39d18b : FAILURE in 38m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7d5f33e4c4364872bb2a2a818d0a855f : FAILURE in 37m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bcdfc552574b4d56a4f33cbbd0603233 : RETRY_LIMIT in 4m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9f67430e787148ceb61566f7a5c75e45 : RETRY_LIMIT in 3m 36s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ad4d218e10e445c08ed742d24bad2345 : SUCCESS in 2h 51m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b5a7562fd7cbeb5ac81fda04952fc0f4b1ce8ca","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-06-17 07:58:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c222e452622acf8c7d3950caf324c3c886a1765a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-17 08:10:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"90b74e2808d6dc12ef40ee9dbf7cb18213a7ab3e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-17 08:34:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b8d686fa377380ee6bd06ddcc88804f3693250b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 10:41:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e8c0303edc340f896f564854933a5fd : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f8aef726565464a8c5df10b4b1a2a18 : SUCCESS in 8m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d23edd46468844e3ae6f387c2b53a91a : SUCCESS in 8m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fc561964c54945da87cef0a13f4bf009 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5200f7959c2d40449f46f0a3e521586d : SUCCESS in 15m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7694b154f1394b45a23a2e69e1b94bc3 : FAILURE in 12m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9bc8fd8f4a984272a0442a9754e71636 : FAILURE in 15m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fbfcb4aa8a7343c6bdf6928ad487859e : FAILURE in 15m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d93bf2b461c744f499f674157a4f53db : SUCCESS in 26m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/26c3f584b0124e5683e2b5fb4d964e9d : SUCCESS in 30m 07s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0ee645252b0d46ee89c1d56fd7a615e2 : SUCCESS in 2h 03m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b54b3d310a84fb54cb4a9252d1f4096ff248a65b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-07-02 11:45:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"969f2829db69d59dd64f958116e7f29e7894e063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 14:02:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debf23b99bd240b78986e6d6fe86e197 : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb4d14e295584e8991416fe25a608097 : SUCCESS in 11m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2ebd77b2bbc6435ba3d5349e41b948bb : SUCCESS in 10m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce492ab549c04dbba05901d9ba455ff3 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65de0aeb9ca64f2b9dc07e6a814ace5c : FAILURE in 14m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/77bcbd11750e4505a0fab8b487d7fd9d : FAILURE in 20m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c0c1c6e862c34d1db3de70e2debfeea8 : FAILURE in 16m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/41c8bde5622d4e72a311f04d2b572aa5 : FAILURE in 15m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2da7bb4f32ef4ede9f90538d7a9aa7d2 : SUCCESS in 25m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/891d7c13394442a6b7015cbab55129bf : SUCCESS in 27m 27s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/273acd71a36748d7ad0f994d4f7271b9 : SUCCESS in 2h 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a0e60d311a2c7706e665b6af431924e1d6f2062","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-07-02 17:41:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"598be8d7fcd7aaf31206fbd22ed09a3952407aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 19:41:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1823a2e38f746d780d286c914a9fe6e : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5af54cbf693d41989b5598b5a97bc7d6 : SUCCESS in 10m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a8495e858ef4143ac877dc6ad4fdc22 : SUCCESS in 11m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c3237e12a5d242f492874c982c80229a : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/825c469f056e43a79f9c76c0bbf2a1bf : SUCCESS in 13m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/81e0aecf96fb4db1b27c7093de5a2c9a : FAILURE in 32m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0f0f0fd1d2204912b82292353f3f2a2c : FAILURE in 34m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6800d7d3d70445f692116cb262cde11c : FAILURE in 31m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2e93185fd8b046a7b1b201f867b85f41 : SUCCESS in 25m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/615151cc51c5434a9d3f97bb80e10dd4 : SUCCESS in 29m 56s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8ffb7ab9a45f4366b2d4a2404d97f87b : SUCCESS in 1h 55m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7d9c37fc3fc6062aea8c74324e2f1e00f059f15","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 12:32:04.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck\n\nClean backport LGTM","accounts_in_message":[],"_revision_number":3},{"id":"822a0a5849b5df356643ad484e6e83aa7441aa54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:08:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/febb493286214f3d9a2a16de715e0e4c : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc0b0670bef241f886366084d95ae4c8 : SUCCESS in 9m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/68674598f8354abc823c76671f695378 : SUCCESS in 9m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/36ac7455b3a448619bf89a3dad69530e : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77fa37709c21469cbd7c65009b13d259 : SUCCESS in 14m 34s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d3633dec78eb405fbfcfb0115f938494 : FAILURE in 42m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/362b693ec9c04494a573e111955fec77 : FAILURE in 38m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c38c1a6c19574a47b031ac69dcff8050 : FAILURE in 35m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5b6f3833fdcf4be18e426168901243fb : SUCCESS in 24m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/725a84afdcf847f0872959e3d9774dcb : SUCCESS in 32m 02s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8e79b561b3c94f1aace43161fdd79aab : SUCCESS in 2h 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b237241dc902c7ad0eaa6c81b5891058ac8857d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-07 05:31:30.000000000","message":"Patch Set 3:\n\n@Hervé, we\u0027re now getting this error in the devstack jobs:\n\n[ERROR] /opt/stack/devstack/lib/swift:848 swift proxy did not start\n\nWe had a different issue (related to wsgi) and it seems to have been fixed now. Do you know anything about the new issue? Is it a known issue being already solved?\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"e9b8cc80141bd509a9a7e77a3f86cdfea47e03ec","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-07-29 11:47:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8e17a4b970e0eb18855dbc976dd83dbf13e8c20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 12:36:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f02e1b0e6044a49c7fc79715e12909 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa0c4c806ffc478bb63631b33086b7fb : SUCCESS in 8m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bd555abbc02e46d798114c18301496e8 : SUCCESS in 9m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/358ef319100d4fc389d18e18d127f4b7 : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2de91ee8cff64a72a3e534c812c96dcd : SUCCESS in 13m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/56f4d0de652f425db554ef4ad46c1555 : FAILURE in 40m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/34df49cdb53d4d879505d36ae20fc8e0 : FAILURE in 39m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/42637748ab014dc79c3394aa4870a49f : FAILURE in 42m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ce26bfd70791473291dd9ded82f51c1a : SUCCESS in 26m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2c90d124d0c149629a6289bebed63202 : SUCCESS in 24m 51s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c6d7bdc088894c52a4e42d118dba7c9f : FAILURE in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"350a6d46b549fc37e66d64dc569d92051a3dbbba","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-09-09 13:22:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"451750f44a3c99ba26e8fe9cedebcac7416d78cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 15:50:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b6d998f70a46e8b7000ec189a2cdc3 : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/208e772ff56c4e0fba1b1511ae84ee5e : SUCCESS in 9m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31895a4e85bf4b9694681795b5fbe3f8 : SUCCESS in 10m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/314e10caf6704846875ca0927e580afe : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04e99adef0bc42669ac43b1a82aa9400 : SUCCESS in 14m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4e01fe14276c43ec9298ca9306f9448a : FAILURE in 28m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0ff3cf4ae2064259af70e9d97b378955 : FAILURE in 44m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1c2ea7f4d32e4f9f85724e1c33b29e26 : FAILURE in 39m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d7144071d2e5490a8e7fd106bbad7cfd : SUCCESS in 25m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/15e8158cede247a2a876eae46e8ba069 : FAILURE in 6m 10s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3eebcba0752f4478a374a416837f43cf : SUCCESS in 2h 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e936700694c7344fb40d327f6048ceede264bd78","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-08-18 08:54:22.000000000","message":"Abandoned\n\nqueens is EOL.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"47736a6d566c173e36ec908c389df2c95350b6d6","revisions":{"5c38ccf169104c396bc4f0870dbc48b36d2a4baa":{"kind":"REWORK","_number":1,"created":"2020-06-10 13:14:07.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/30/734830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/734830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/734830/1"}}},"commit":{"parents":[{"commit":"da0e2639a8d89e95ac3011d0c21696d37b03301e","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da0e2639a8d89e95ac3011d0c21696d37b03301e"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Cédric Jeanneret (Tengu)","email":"cjeanner@redhat.com","date":"2020-06-10 13:14:07.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n(cherry picked from commit a25c8fab881decff0e61deffe1895f1d99132f9f)\n(cherry picked from commit 4d3ef8887fb765c9edd36f62cc66904aff4c1b0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c38ccf169104c396bc4f0870dbc48b36d2a4baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c38ccf169104c396bc4f0870dbc48b36d2a4baa"}]},"branch":"refs/heads/stable/queens"},"f409dcf484a58f2cdd5ae6988a32d42f888d2296":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-17 08:34:06.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/30/734830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/734830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/734830/2"}}},"commit":{"parents":[{"commit":"c75d5671217c51d30fdf30a134ee8d5afd8212ad","subject":"Add bindep.txt with the CI dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c75d5671217c51d30fdf30a134ee8d5afd8212ad"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-17 08:34:06.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n(cherry picked from commit a25c8fab881decff0e61deffe1895f1d99132f9f)\n(cherry picked from commit 4d3ef8887fb765c9edd36f62cc66904aff4c1b0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f409dcf484a58f2cdd5ae6988a32d42f888d2296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f409dcf484a58f2cdd5ae6988a32d42f888d2296"}]},"branch":"refs/heads/stable/queens"},"e181f52e826df7b79681f077417b137a977c0174":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-02 17:41:27.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/30/734830/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/734830/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/734830/3"}}},"commit":{"parents":[{"commit":"fdca58490fe7f74f534e7d3132697eb4449473fa","subject":"Add bindep.txt with the CI dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdca58490fe7f74f534e7d3132697eb4449473fa"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Adriano Petrich","email":"apetrich@redhat.com","date":"2020-07-02 17:41:27.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n(cherry picked from commit a25c8fab881decff0e61deffe1895f1d99132f9f)\n(cherry picked from commit 4d3ef8887fb765c9edd36f62cc66904aff4c1b0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e181f52e826df7b79681f077417b137a977c0174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e181f52e826df7b79681f077417b137a977c0174"}]},"branch":"refs/heads/stable/queens"},"47736a6d566c173e36ec908c389df2c95350b6d6":{"kind":"NO_CHANGE","_number":4,"created":"2020-07-29 11:47:24.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/30/734830/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/734830/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/734830/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/734830/4"}}},"commit":{"parents":[{"commit":"efa792da1b5890087a7c8c256cdd92ba435c9256","subject":"Add bindep.txt with the CI dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/efa792da1b5890087a7c8c256cdd92ba435c9256"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Cédric Jeanneret (Tengu)","email":"cjeanner@redhat.com","date":"2020-07-29 11:47:24.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n(cherry picked from commit a25c8fab881decff0e61deffe1895f1d99132f9f)\n(cherry picked from commit 4d3ef8887fb765c9edd36f62cc66904aff4c1b0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47736a6d566c173e36ec908c389df2c95350b6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47736a6d566c173e36ec908c389df2c95350b6d6"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
