)]}'
{"id":"openstack%2Fmistral~735103","triplet_id":"openstack%2Fmistral~master~I5e0fee3a85ff7eeacb2503295da70abbfcf62b49","project":"openstack/mistral","branch":"master","topic":"py37-victoria","hashtags":[],"change_id":"I5e0fee3a85ff7eeacb2503295da70abbfcf62b49","subject":"Update Zuul job configuration","status":"MERGED","created":"2020-06-11 10:13:04.000000000","updated":"2020-06-15 18:00:37.000000000","submitted":"2020-06-15 17:55:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735103-1592243719415-6db76d6a","meta_rev_id":"9cfedbbb2352589709bd2d2c3404ccbff8d7d721","_number":735103,"virtual_id_number":735103,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-15 14:41:37.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 17:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-06-11 10:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-06-15 05:37:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 05:37:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-06-15 17:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a02fff6ac33a4b55c5d987e7e6065a3f2b9c9f00","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-11 10:13:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69ff75b5cf58c93595faf8332ed2134b6312bdb9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-11 10:14:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73161c06686b6a825f6cc1514c5dbde6f432863a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-11 10:18:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aeb8f7d6d4ff7556af4f35176286911d7bc1d427","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-11 10:18:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c03be4a43606b673599b143a44b9a7ed520e6fe9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-11 10:26:34.000000000","message":"Patch Set 4:\n\nLooks OK. Just one question: I see that these change lead to lack of Py37 in the \"check\" queue. Is that OK? I thought we were supposed to be testing all Py3 versions starting Python 3.5. I might have missed some relevant discussion..","accounts_in_message":[],"_revision_number":4},{"id":"c62449b7234464665fd4366b738b73c8a8a01737","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-11 10:30:25.000000000","message":"Patch Set 4:\n\nVictoria template only tests py36 and py38, see https://opendev.org/openstack/openstack-zuul-jobs/src/branch/master/zuul.d/project-templates.yaml#L456\n\nas well as:\nhttps://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nSince we switch to Focal, no need to maintain py37 anymore","accounts_in_message":[],"_revision_number":4},{"id":"ea2b36cb3b26aec3fcdf859e8e0dcebe583d72a3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-11 10:35:38.000000000","message":"Patch Set 4:\n\nOk","accounts_in_message":[],"_revision_number":4},{"id":"5a2e1c4187155d29de8500c9d13409103ee1f4f5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-11 10:36:35.000000000","message":"Patch Set 4:\n\nThanks for clarification.","accounts_in_message":[],"_revision_number":4},{"id":"7c185254da82983e0ff0e5bfb8d9fcca19ea709b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-11 10:36:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"eba772579b270a1ffd9db4b06e526beef3b17df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:18:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429dae2a7f034808b1262d5abb7e366b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0497fef474d4f279e7bad6e26a8fe34 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bada44ba0400407d91b94b9142221c47 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b68328ea664d398188edabb8560bc6 : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39bd9356dc1f4cb6bbfdfba8e6047208 : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68946efdb7c54409b36ce8263790d9d1 : SUCCESS in 8m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3022aaff6bc1423997aff28ab6774415 : SUCCESS in 41m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76dd7da91a4942e7a02dc6b51a0a2917 : SUCCESS in 49m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b0ccdde95a24c3782924d74f666ae8a : SUCCESS in 46m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b79e6968010d4baabd2488a604bfdefb : SUCCESS in 41m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8add3ad519c646d3b082be8eb2e81ced : SUCCESS in 29m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/da917c2bb86c4c8aa5341e887dd7a993 : SUCCESS in 29m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/95735b5878ef41dba6a908baade10d02 : SUCCESS in 27m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7bba63e89d474a83816b730bfd11bf92 : SUCCESS in 1h 54m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d02c16d1b84341154ea8085c20747185b61c4d9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-15 05:37:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a23b9155dc745b0975b4dfd60bb7dc728259d59e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 05:37:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c4a17b8acaba142f84f71c07b4c12732046e4a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 06:11:45.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d5c3bbb2b0344c78dca93ddc1704d24 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eb5ab0f3cf642b880ebd71abb6f9542 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52abd797f55444f784c812668cfefa7d : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b7756574af48858e808219ddee1ca9 : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61ef1e6ab3964917b36411dab529d01a : SUCCESS in 9m 27s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/063a0e7595534c8aadb2f53202a6f4a3 : FAILURE in 12m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51ba94091aac4a99b66778f121963d6f : FAILURE in 14m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/386a0fe437804b5a9ee732659f753a35 : FAILURE in 16m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c7f47aa6cec455db3d1c8f90cb1b401 : FAILURE in 15m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a490a309b788431fa27f59de305eb709 : FAILURE in 14m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8cd769a016f44e219b2b39162c9f0b1d : SUCCESS in 33m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d793fc32f7c49ff88037013deb1bc32 : FAILURE in 30m 35s","accounts_in_message":[],"_revision_number":4},{"id":"591b6e0dd3c66d4f01fc7f108551f16d2405eaa6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-15 06:37:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"af230e77312bb8ec3b1978968325f145b368b29c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-15 06:37:20.000000000","message":"Patch Set 4:\n\nThe failures look unrelated to the patch..","accounts_in_message":[],"_revision_number":4},{"id":"ffecc13fa2455611decfc55962bf8a12ed21ddec","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-15 06:37:55.000000000","message":"Patch Set 4:\n\nRenat, recheck won\u0027t help - uwsgi was updated and this needs a devstack change first, see https://review.opendev.org/735494","accounts_in_message":[],"_revision_number":4},{"id":"1a1630982b542454a966813295c4777e52ddeaed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-15 06:56:01.000000000","message":"Patch Set 4:\n\nOk, got it. Thanks Andreas.","accounts_in_message":[],"_revision_number":4},{"id":"6dcdbd40a6fe2b00d4e65541453db66e2272ef75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 07:20:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71227e10b584ec5a8a350af7a061022 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a5c129f61f243a2869b913cb2ce59bb : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9b737921a004db1a809ff1576502ddc : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6432344667400e8256aae32f387996 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f513b857c1f49d9a1aa010c1af79ea7 : SUCCESS in 13m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f73e474423147069bd42ad67eee0d73 : SUCCESS in 11m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/46c5c843cc4242bbb6d760fb8b408f87 : FAILURE in 18m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f4fe250fc9d41cf97fb79b568356b7e : FAILURE in 18m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eed807734eb64cc183862c3325d4c76c : FAILURE in 23m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/16eba5b59b4f4172851371a8485a44d0 : FAILURE in 21m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d40e9f40084f4e66ade043d219c45426 : FAILURE in 21m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/de101dfe2f9e44f285e545a86f2200f3 : SUCCESS in 31m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e34ccf71496548b793c1096476c312fa : SUCCESS in 31m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/59bdb70caf3449f18608d3920c8a99c8 : FAILURE in 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"334e3c6172f189f2127d659412fb3e0dce8ae5ca","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-15 11:57:46.000000000","message":"Patch Set 4:\n\nrecheck - devstack is fixed","accounts_in_message":[],"_revision_number":4},{"id":"737053d5abc7b2e337da77b4da4d803ff4e0c626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 13:47:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb340d4e00d4c5581692d5908963449 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb3764412eee4882a398a695d7a2b52b : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/687d55f026f1486b8b8765984d6d3eca : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76917738f38240c7b9769504f5db9f5f : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337663d12b054afba1fe86d04c2f0291 : SUCCESS in 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b01d129f24c4809953c0de891852f89 : SUCCESS in 11m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d4ccbf12c3ce4c8ca45aa11b4bb30d65 : SUCCESS in 36m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f21e462f4ab48de9033c73b08824822 : SUCCESS in 47m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c50c1beb234049dda414144130c7f47c : SUCCESS in 54m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/029e3e9cedef41de9aaa22e98c2adb87 : SUCCESS in 46m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a67c99b23b504f0e8ab822402a9d31b7 : SUCCESS in 56m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6d75c290fd21422cb5f35f4c863c27e8 : SUCCESS in 33m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0b4ef5e4000c409e894e5a81a665fade : SUCCESS in 29m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/91b2ac9579784344ac92fcf4be979c12 : SUCCESS in 1h 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"697e200bd28c21dcc48287db83ff4999e0a77674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 13:47:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7bf706f31a26550e529bb8cda18ece63e88b0d06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 14:36:39.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/021768858ba84398a31b90984a43b503 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61735fd4213b473696787785ad1b8579 : SUCCESS in 12m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6362512bcf704098bffbc56fa988f4e2 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd970c711f4420489c8870739379e8e : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5408423176ef415986e4a373d441c03f : SUCCESS in 12m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/be553ed878024135801b4de9a61305c3 : FAILURE in 39m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2fcbd62a7094b34a2f11b9e7e4bfe0d : SUCCESS in 43m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aad8124e1b3b4716a43de69efefea0b3 : SUCCESS in 46m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/37dc17fe10384b29bb1011e1f017264c : SUCCESS in 46m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/53ba1305b3984f87af63d0e74074c397 : SUCCESS in 45m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fc8f33e7dd944a3981ec6294e95f5695 : SUCCESS in 33m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0fc255212d53463c969e22055eb3de1f : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":4},{"id":"9116b716d3d6c31394572dff4f829a89a9f5a447","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-15 14:41:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4aa973fa68aa086827d8a02ccac4eaad7e432827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 16:47:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf205e55b3db4c9cbaaac0874522ae92 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0805f040d60c4bdf9cdfd861f64e0f76 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2936cbd91f945d5b3cc4aee009230be : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95f1f4124d04958a0e56fc88b297807 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcd5d3b471b3475e8c5e27617502d453 : SUCCESS in 10m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f51abf28a844968b1304e1b0d77fa52 : SUCCESS in 11m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dcb56bad9d40491fb492ab05ef8b0a4e : SUCCESS in 49m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/200ea39a98af46c4a526c1eeb53fb2bb : SUCCESS in 39m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9162a27d0ef94825a26b0c79f7de027c : SUCCESS in 45m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/942720c487da427dbb806f9cb5c22568 : SUCCESS in 50m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a50406efb16548c5a24ca97f1669d232 : SUCCESS in 51m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/841df741a6ac417dab845975a041b839 : SUCCESS in 33m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a2cc25ad08fd49729a0e116c495cbe13 : SUCCESS in 32m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/513b0d9467d044da9462584eac68fa3a : SUCCESS in 1h 56m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"19d94aed312efd167387a84f977b2aa8c41d3326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 16:48:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8b26023d4da99c026522e3ec9998caa74fd13acd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:55:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cb5b648dd5fca45e945e75a47b5800e354e3a81d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:55:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc03f73062141aeb59e2205348dfc2c : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79a1c0e705ed4840a5f608a3f7e7a999 : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d21845f527c4bf38d09a2475d4349be : SUCCESS in 17m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43cb580e7104675846db8867d749059 : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3345fbe5f0144ebb8f82e5992f6e863f : SUCCESS in 11m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e8663625dd8e49bb8a33de053bbdd77e : SUCCESS in 50m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3289b51e27440e5bda97b330ed41c3e : SUCCESS in 41m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78ec87cd15fe4d2484836258ede9b101 : SUCCESS in 49m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/527dcbbfd4634f5087ddfeded677fc95 : SUCCESS in 44m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3961643e1d254fdab2967e32ac8f3c20 : SUCCESS in 1h 04m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/74f8ba836d3d498793a814b017d99df0 : SUCCESS in 37m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bace18d351f24cf8a772fb5d4ea66f30 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":4},{"id":"6c1219461f781e2c1c2bf4364c2f49e63e43a686","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:00:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0cf345aecc4441838e707843ced4ac : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"104ab8e9ac8b7b5639b830069206db407f329800","revisions":{"3c164e13ab579ec1b5b85a3c6a229209c12180b1":{"kind":"REWORK","_number":1,"created":"2020-06-11 10:13:04.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/735103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/735103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/735103/1"}}},"commit":{"parents":[{"commit":"d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:11:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:11:41.000000000","tz":120},"subject":"Remove required-projects from check only","message":"Remove required-projects from check only\n\ncheck and gate queue needs to test the same environment. Remove\nthe required-projects from check queue.\n\nThis removes also the jobs from the Zuul, they are part of the existing\ntemplates and are still run.\n\nChange-Id: I5e0fee3a85ff7eeacb2503295da70abbfcf62b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c164e13ab579ec1b5b85a3c6a229209c12180b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c164e13ab579ec1b5b85a3c6a229209c12180b1"}]},"branch":"refs/heads/master"},"275e56aefaa339fbba7095c4f6a6003b54e99467":{"kind":"REWORK","_number":2,"created":"2020-06-11 10:14:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/735103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/735103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/735103/2"}}},"commit":{"parents":[{"commit":"d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:11:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:13:41.000000000","tz":120},"subject":"Remove required-projects from check only","message":"Remove required-projects from check only\n\ncheck and gate queue needs to test the same environment. Remove\nthe required-projects from check queue.\n\nAdd openstack-lower-constraints-jobs to templates.\n\nThis removes also the jobs from the Zuul, they are part of the existing\ntemplates and are still run.\n\nChange-Id: I5e0fee3a85ff7eeacb2503295da70abbfcf62b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/275e56aefaa339fbba7095c4f6a6003b54e99467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/275e56aefaa339fbba7095c4f6a6003b54e99467"}]},"branch":"refs/heads/master"},"e226b5485f5a99e8ab85bc268d42117219f36d88":{"kind":"REWORK","_number":3,"created":"2020-06-11 10:18:04.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/735103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/735103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/735103/3"}}},"commit":{"parents":[{"commit":"d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:11:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:17:41.000000000","tz":120},"subject":"Update Zuul job configuration","message":"Update Zuul job configuration\n\ncheck and gate queue needs to test the same environment. Remove\nthe required-projects from check queue.\n\nAdd openstack-lower-constraints-jobs to templates.\n\nThis removes also the jobs from the Zuul, they are part of the existing\ntemplates and are still run.\n\nRemove tripleo-ci-centos-7-undercloud-containers job, change\nIe94e03bdcc14918eb4095e9581d4a935a59b0d6b changed that to\ntripleo-ci-centos-7-containers-multinode but missed the gate one. Since\nit\u0027s non-voting, remove it from gate.\n\nChange-Id: I5e0fee3a85ff7eeacb2503295da70abbfcf62b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e226b5485f5a99e8ab85bc268d42117219f36d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e226b5485f5a99e8ab85bc268d42117219f36d88"}]},"branch":"refs/heads/master"},"104ab8e9ac8b7b5639b830069206db407f329800":{"kind":"REWORK","_number":4,"created":"2020-06-11 10:18:34.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/735103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/735103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/735103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/735103/4"}}},"commit":{"parents":[{"commit":"d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0d75be7cbeca4b4445b0799f196feb5ddeb6b1b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:11:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-11 10:18:28.000000000","tz":120},"subject":"Update Zuul job configuration","message":"Update Zuul job configuration\n\ncheck and gate queue needs to test the same environment. Remove\nthe required-projects from check queue.\n\nAdd openstack-lower-constraints-jobs to templates.\n\nThis removes also the jobs from the Zuul, they are part of the existing\ntemplates and are still run.\n\nRemove tripleo-ci-centos-7-undercloud-containers job, change\nIe94e03bdcc14918eb4095e9581d4a935a59b0d6b changed that to\ntripleo-ci-centos-7-containers-multinode but missed the gate one. Since\nit\u0027s non-voting, remove it from gate.\n\nChange-Id: I5e0fee3a85ff7eeacb2503295da70abbfcf62b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/104ab8e9ac8b7b5639b830069206db407f329800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/104ab8e9ac8b7b5639b830069206db407f329800"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
