)]}'
{"id":"openstack%2Fmistral~736158","triplet_id":"openstack%2Fmistral~stable%2Ftrain~Icf460fea113e9279715cae87df3ef88a77575e04","project":"openstack/mistral","branch":"stable/train","topic":"eyalb/yaml-stable/train","hashtags":[],"change_id":"Icf460fea113e9279715cae87df3ef88a77575e04","subject":"Disable the use of anchors when parsing yaml","status":"MERGED","created":"2020-06-17 11:02:10.000000000","updated":"2020-06-17 15:53:33.000000000","submitted":"2020-06-17 15:50:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736158-1592409044034-2c7c96e1","meta_rev_id":"867448d998775a16de46b84e0da81a32f2f6fe45","_number":736158,"virtual_id_number":736158,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 15:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-06-17 12:15:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-06-17 11:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2020-06-17 11:18:38.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2020-06-17 11:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-17 11:02:10.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 11:02:10.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-06-17 11:18:38.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-17 12:15:13.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-06-17 15:50:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f5a5608383f8390d92a9e8e728e7565bc1230b9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-17 11:02:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df0ee42f5c6ca56d1b157b8ea9b0e8c1e274ccbb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-17 11:53:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c10c4ddb386c006b399dea258f92b3ac3b7ce7a5","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-06-17 12:15:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c507514571c108c00e19b7df04ade67a1f4e2fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 14:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d8225df2504d91a41ad8b0571a5faa : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad35f01ed5964a4fb05c7b705f327802 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d029e526cc2548f1ba0d3e0f46c44d2e : SUCCESS in 10m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e48cf2bc4ba34c07ad10b7d48a9650ef : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2abb5394d7d46bbac56c2841dd02756 : SUCCESS in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a84d68a431e74bbcac4d6529a28dccbd : SUCCESS in 9m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6173fe501e21425783257f486ea5132e : SUCCESS in 54m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2d525afcef1b41debb70243ddb7cfa67 : SUCCESS in 42m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/23a62b6472c944efb97466fadc712157 : SUCCESS in 46m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b7c1a0c44baa4440bb794e71440de6c1 : SUCCESS in 31m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/653ba42eaf40468096590dcb65da102e : SUCCESS in 29m 58s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/fa532f1e1c424898ac641ec286389b99 : SUCCESS in 1h 25m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/547a818f5fb14743a08b2694d9aa82fe : SUCCESS in 2h 57m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/64fc683854af41cf832377d671531922 : FAILURE in 33m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e4f2ae248224dbc8ae530d3862faeb3 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"261a6a39a1e65c1b8c125658aa6ad600036e4e94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 14:01:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cac86852019962231b8034546d8489d63dd4216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:50:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d0842533a4a4b4bb0fdeb50134f7558 : SUCCESS in 9m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42afd2313ac74e3daa338c7e12670b99 : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51eb44b8ff454d94ac1c7a2bba3adc9c : SUCCESS in 12m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12f1f08c53b9481fb34c7e95772ea92f : SUCCESS in 13m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4535fc80986f4d1b9718829f7d05438b : SUCCESS in 9m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bd45b1f48fd74290be6d4e951954ffa0 : SUCCESS in 47m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ca823fbf9f3a4668b6c0281930b29624 : SUCCESS in 43m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1c6ce31f4fdf4b47888595e8e94fecbb : SUCCESS in 32m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/924dc0de2d5041ff8d777a4986806767 : SUCCESS in 33m 07s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cdc179b6b4de4571b7117513724f7a7d : SUCCESS in 1h 43m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2d8cf528a8e941a7949605999377b679 : SUCCESS in 43m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cc40c494a53444086c0c208a1e1979f : SUCCESS in 12m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fa5ea09d4b924ff6aa10b35c687e425b : SUCCESS in 1h 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"95ca8dfaf261f1c7976d8340b9f46485681a2ff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:50:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd2925eebbac19bb5a19fc439d61dc98d4859157","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:53:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b170b62b3244b6b950810c5cb30564 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0824730fcc4bed3614071be2a1fc82a0ed29114","revisions":{"f0824730fcc4bed3614071be2a1fc82a0ed29114":{"kind":"REWORK","_number":1,"created":"2020-06-17 11:02:10.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/58/736158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/58/736158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/736158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/736158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/736158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/58/736158/1"}}},"commit":{"parents":[{"commit":"30039178714ea776df2b60a0f49427c239eb6d22","subject":"Merge \"Add YAQL sanitizing for iterators\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/30039178714ea776df2b60a0f49427c239eb6d22"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-17 11:02:10.000000000","tz":0},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n(cherry picked from commit eac23d9e774f658f9d4743c99aa2743eb104c3f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0824730fcc4bed3614071be2a1fc82a0ed29114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0824730fcc4bed3614071be2a1fc82a0ed29114"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
