)]}'
{"id":"openstack%2Fmistral~736163","triplet_id":"openstack%2Fmistral~stable%2Frocky~Icf460fea113e9279715cae87df3ef88a77575e04","project":"openstack/mistral","branch":"stable/rocky","topic":"bug/1785657","hashtags":[],"change_id":"Icf460fea113e9279715cae87df3ef88a77575e04","subject":"Disable the use of anchors when parsing yaml","status":"MERGED","created":"2020-06-17 11:15:51.000000000","updated":"2020-06-18 12:39:36.000000000","submitted":"2020-06-18 12:39:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736163-1592483976290-e21aa85b","meta_rev_id":"c8e49cc07dfdea811cc6a56472a9b520c46bed08","_number":736163,"virtual_id_number":736163,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 12:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-06-18 08:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2020-06-18 08:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 11:15:51.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-06-17 11:19:05.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-17 12:14:34.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-06-18 12:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ce02873d040000b5d8287e621b4da323d606343","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-17 11:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92b5fc709e9741bebee00c459f275aaff7971396","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-06-17 12:14:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4e77ed687059b093502ba64694dbe09684783f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 13:08:42.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c84402255d4a19a9b55d51fff106e0 : FAILURE in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f88dc42cdb0d4c49a8e8497666d07ff8 : SUCCESS in 9m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3df39687f6a44824867632c1db1c514e : SUCCESS in 9m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1f92ff6de07640689272a125841ebba0 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4b3ab9d0de94684966bae582b33596e : SUCCESS in 10m 20s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f60a920ca5714ab2bc7b0fb51d3a6539 : FAILURE in 9m 47s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f9c65fe5929345fcb7c9fded45b8a78b : FAILURE in 11m 40s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a5cd207cb1f24a3f8a2b3597e6f09e8e : FAILURE in 8m 54s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ee5b955fd596486d99dba676ca66058f : FAILURE in 27m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d218f1465494f218212cd0ba1a3ac06 : FAILURE in 27m 20s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/946111581ab44b09976429f65d0bd775 : SUCCESS in 1h 51m 37s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9086d72076aa434a86419127a1c220b2 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9e97dc55fa245b18658fc9977ee08dd : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c72e83f250e4fd446873ee657bfa83c257dd0ce9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-18 07:47:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a31ea7831b2a40a9e0472a36ef761d1e1f67ff8e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-18 08:10:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2372aff1e963e69705ed25ff12f9e407b27a8845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2954b2ca034761bba5d1738d889159 : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3eb4aaf05366490a980af2a67b4e420a : SUCCESS in 12m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e7f35348340413ebaa66a6f5cbc7a5e : SUCCESS in 14m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/40eba220caa5430a9bdeb1b30cbab7cf : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21f1a365fcc47f482996e8dd7383233 : SUCCESS in 14m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/22a98011c2eb4382b99a8903a2cdd2d6 : FAILURE in 18m 34s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dad471af11b94898bc1c310d2b53e93e : FAILURE in 19m 02s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fdf180d0f04a4a9f990f9e3f6f3e8c26 : FAILURE in 20m 13s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9667e6e9c5bf4a7ea1956f7f8bc41081 : SUCCESS in 29m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4aa4601212bd4348a7bb64a66ef83e23 : SUCCESS in 27m 40s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3c78d5dd4ed94abca84807f6f1c95397 : SUCCESS in 1h 54m 08s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/38e8bd6f9d9d4525b9e1c4cf095f2919 : FAILURE in 12m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13cf9fe172624127b3259efea59701e8 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":2},{"id":"bcf693155b7c65a9700ccbb6bd314bbbf19d86c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:49:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57a6814c48584e4b62963927c40bcf97aff98e6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:39:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d8fd2196ced83acb6dd4ef6002cecd0517b11a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:39:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fb0c6b1233c4c93849467079d4037a9 : SUCCESS in 10m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ab0cafcbd294e2fb3149f9238b30da2 : SUCCESS in 12m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29dcb86ab1074b38bcb4c4fec52d0149 : SUCCESS in 11m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7d59b53c283b46ab8564bb8123dbf293 : SUCCESS in 8m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bb195569a7d74d75a57f95b35fcb09f0 : SUCCESS in 28m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/462514a95a7c4dc8ba8b8068474dd3c1 : SUCCESS in 30m 07s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/40d77a2578ff4c38878ac42cd0d5dbbd : SUCCESS in 2h 43m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3871cf1bb38043a79f4ae9ca6b830843 : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2968b8bbb743ada6df7e36c969228bef42830224","revisions":{"679e21a46f2874698c093bebcabb3e612d5a6f54":{"kind":"REWORK","_number":1,"created":"2020-06-17 11:15:51.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/63/736163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/736163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/736163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/736163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/736163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/736163/1"}}},"commit":{"parents":[{"commit":"4ad66f140bc7cce9c1c7805a22d9d3c5600642d6","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ad66f140bc7cce9c1c7805a22d9d3c5600642d6"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-17 11:15:38.000000000","tz":420},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/679e21a46f2874698c093bebcabb3e612d5a6f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/679e21a46f2874698c093bebcabb3e612d5a6f54"}]},"branch":"refs/heads/stable/rocky"},"2968b8bbb743ada6df7e36c969228bef42830224":{"kind":"REWORK","_number":2,"created":"2020-06-18 07:47:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/63/736163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/736163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/736163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/736163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/736163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/736163/2"}}},"commit":{"parents":[{"commit":"4ad66f140bc7cce9c1c7805a22d9d3c5600642d6","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ad66f140bc7cce9c1c7805a22d9d3c5600642d6"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-06-18 07:46:51.000000000","tz":420},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2968b8bbb743ada6df7e36c969228bef42830224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2968b8bbb743ada6df7e36c969228bef42830224"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
