)]}'
{"id":"openstack%2Fmistral~740622","triplet_id":"openstack%2Fmistral~stable%2Fussuri~Ic46d3d711a03fbd6364809a0fddcbb48f3e28d2a","project":"openstack/mistral","branch":"stable/ussuri","topic":"bug/1887335-stable/ussuri","hashtags":[],"change_id":"Ic46d3d711a03fbd6364809a0fddcbb48f3e28d2a","subject":"Add a migration that deletes redundant delayed calls","status":"MERGED","created":"2020-07-13 07:15:15.000000000","updated":"2020-07-15 10:02:27.000000000","submitted":"2020-07-15 10:00:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740622-1594807243509-56289871","meta_rev_id":"4e966015166e9cf519c9c33a870930c23e39613c","_number":740622,"virtual_id_number":740622,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-15 10:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2020-07-14 12:01:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-07-13 08:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2020-07-14 12:58:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-13 12:04:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-07-14 16:16:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-07-15 09:03:46.000000000","_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"date":"2020-07-13 07:15:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-07-15 09:03:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-13 07:15:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-13 12:04:57.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-07-14 12:01:04.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-07-14 12:58:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-14 16:16:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-07-15 09:03:46.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2020-07-15 10:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40e24dd36e7b55ad9fc621bf8f89ba37e5a910d7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-13 07:15:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a3a55eb9521526800398221a8da9d570b88c2546","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-13 07:17:02.000000000","message":"Patch Set 1:\n\nHi,\n\nIs it OK to backport this patch to Ussuri? It adds a migration but the migration doesn\u0027t alter the DB schema. It just cleans up a table.\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"1611c443a7971d64f64938b919658c75b2cdbc5b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-13 07:53:57.000000000","message":"Patch Set 1:\n\n@Thierry, ^","accounts_in_message":[],"_revision_number":1},{"id":"31d129ab93ac1a7cd1be3d405fd242dcc51cd0ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 08:16:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe08d44046648a982bf1ad638df5c29 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/280dbb326f7c4fcd933f88bc9f9b6a25 : SUCCESS in 12m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e22a53d2399048339531dfeb8f22685f : SUCCESS in 13m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b7661b16b2d406a94b22d7eac495c42 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88e0e20bc89400aa387a53879aee8fa : SUCCESS in 9m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5f83b4f5fe4372a7fc931412837724 : SUCCESS in 13m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fdbec44c93f6486dbced5900e040303a : SUCCESS in 58m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0369fc9ae4bb4ec8b2f0c9d7954a0013 : SUCCESS in 58m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/598e6cda0342444ab92a4f8320dda766 : SUCCESS in 58m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8c0c7403b98f4ee78da9157c6ad03562 : SUCCESS in 53m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e64589e8e35b400bbffbfa9c152d9f58 : SUCCESS in 46m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b960c0111f96450895d40d61a10e999a : SUCCESS in 31m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f8962dc53ced497a9e42890e988337c5 : SUCCESS in 29m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/14014df6767d426eb4c37543ccc2a35f : SUCCESS in 57m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60c07cac304647ecb2d0ed28e8a2067a : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3b4b4d5bb6b780c20f170b02b4d181cb6dba112e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-13 08:18:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eaf326f2e66cccc05056af351f58106a2458243c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-13 08:37:22.000000000","message":"Patch Set 1:\n\nRenat, I can\u0027t help you with your question.","accounts_in_message":[],"_revision_number":1},{"id":"1f284944a2968ae3ed00988124de1b267c2f5cd4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-13 12:10:46.000000000","message":"Patch Set 1:\n\nAdding Sean and Tony to the discussion.\n\nWe generally don\u0027t add/fix migrations in stable branches because there is just no way to force people to run them on upgrades. While it\u0027s clear that DB migration is a step when upgrading major releases, it\u0027s not a step when upgrading minor or bugfix releases. So you end up with people running the new version, but not necessarily with the latest database.\n\nThe alternative is to only ship that migration in victoria, which means that people will have to live with the extra rows until then. Since most users won\u0027t run the migration until victoria upgrade anyway, I\u0027d argue the end result is the same, and does not require bending the rules...","accounts_in_message":[],"_revision_number":1},{"id":"8b05861f769554b8702ae2ce907963fc0ef39149","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-13 15:46:48.000000000","message":"Patch Set 1:\n\n\u003e Hi,\n \u003e \n \u003e Is it OK to backport this patch to Ussuri? It adds a migration but\n \u003e the migration doesn\u0027t alter the DB schema. It just cleans up a\n \u003e table.\n \u003e \n \u003e Thanks\n\nDoes master (victoria) have this change as well? If someone upgrades from a release without this change, we need to make sure they get to the same state.\n\nWe do rarely backport migrations. It\u0027s not clear to me that this is a big enough issue to do that though. It\u0027s best to avoid it if possible.","accounts_in_message":[],"_revision_number":1},{"id":"48aa421f80555ff241aabb827e44c876bc191218","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-14 04:00:50.000000000","message":"Patch Set 1:\n\nSean, yes, Victoria has it. No issue here. I don\u0027t quite like backporting it just because it\u0027s not 100% aligned with the rules. However, some of our users asked us about a backport because otherwise they would have to do the same manually (run an SQL script on their DBs). So this is just for their convenience. The point is that these redundant rows in DB may affect performance badly. So if you are not strictly against it I\u0027d ask for your permission to land it.","accounts_in_message":[],"_revision_number":1},{"id":"ae4a9030b77ac90d7607203ee9f86f9e60b3eae5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-14 09:03:37.000000000","message":"Patch Set 1:\n\nSo, the goal is to work around performance issues without waiting until victoria is out, and beyond applying the upgrade, the user has to know about the issue, to take extra manual steps to get rid of the issue (run the migration, despite the upgrade being a stable release).\n\nI feel like giving those users a tool to get rid of the extra rows (or instructions on how to clean up their database) would be equivalent. It would avoid unexpected corner cases due to backporting a migration, or creating a weird precedent for a stable release update instructions.\n\nBasically, if we assume the user has to know about the issue and be willing to take extra steps anyway, I\u0027m not sure telling them \"this stable release exceptionally requires a DB migration\" is the best way to help them through the change. If documenting the query to run is not user-friendly enough, maybe creating a separate CLI tool for the user to run would work...\n\nThoughts?","accounts_in_message":[],"_revision_number":1},{"id":"608c6de903f86b4aae2193184d38fabd29ab317e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-14 10:43:05.000000000","message":"Patch Set 1:\n\nUser awareness is an argument, I agree. For sure we\u0027d have to tell them \"you have to take the latest Ussuri and run migrations again to fix it if you\u0027re experiencing performance issues\".\n\nYes, we can give them just an SQL script, of course, but I just feel like a more friendly tool already exists and it\u0027s the DB migration mechanism that can be triggered just by a simple command \"mistral-db-manage upgrade head\" and it\u0027s always available on the system. So they don\u0027t have to log in to mysql/postgres cli etc. Well, that\u0027s not too difficult too but.. I don\u0027t know :) I\u0027d prefer to keep all DB manipulations behind more high level interfaces rather than letting user peek into our internals.\n\nBut ok, if you don\u0027t think it\u0027s a good idea then we\u0027ll work around that, no problem. I kinda understand that I\u0027m asking something not 100% legal maybe so that\u0027s OK if it\u0027s not allowed.","accounts_in_message":[],"_revision_number":1},{"id":"dd1e854df900b5f13850afe41e027412aff71ec3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-14 11:59:38.000000000","message":"Patch Set 1:\n\nThat\u0027s just my take on it, would welcome others opinions :)","accounts_in_message":[],"_revision_number":1},{"id":"a50b6a02c42468016ddfaf157d06aed5c1c5daf2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-07-14 12:01:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6f719a00d8e50b3b7782b816ac48aa181cbeeda","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-14 12:58:14.000000000","message":"Patch Set 1:\n\nSo let\u0027s say U is the version that is out there today. It may have some performance issues, depending on the size of the deployment.\n\nU+1 can include this that fixes those issues, so it is beneficial for those users that have the problem. For those that have not experiences or noticed the problem, nothing changes.\n\nFuture V users will migrate, and the problem should definitely go away by then.\n\nSo the worst scenario is when someone has the performance problem and upgrades to U+1, but doesn\u0027t know to run the migrations again to get this update. In which case, they are in no worse shape than they were before. And if they come to the community asking why their performance is bad, they can be pointed towards doing this extra step and get fixed up.\n\nI have mostly used sqlalchemy-migrate and not alembic so far, so that\u0027s the part I\u0027m not 100% clear on. I would say, as long as if they had upgraded to U+1 and done the migrations again, then are able to upgrade to V and do the migrations again and have no problems having both U+1 and V migrations run, then it sounds pretty safe.\n\nAdding a new tools, while simple, still seems like a bigger risk. So I think that is an option, but assuming my understanding from above is correct, adding a migration that someone might not know to run seems like less of a risk than introducing something entirely new. Or having an operator manually perform SQL statements on their DB.","accounts_in_message":[],"_revision_number":1},{"id":"1530b2f5d49ac22f4d971eca83ccfa8ca30afd04","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-14 16:16:21.000000000","message":"Patch Set 1:\n\nYeah, still trying to see if there would be a corner case, but the fact that the code works whether the migration has run or not sounds like it should make it safe...","accounts_in_message":[],"_revision_number":1},{"id":"630fc103c3366de912ccb7abdc3319d21a999d64","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-15 08:27:35.000000000","message":"Patch Set 1:\n\nSean, yes, your understanding is absolutely correct. Then I\u0027m going to approve this patch. \n\nThanks a lot for this discussion!","accounts_in_message":[],"_revision_number":1},{"id":"a1e1424c3591190d0948e866c56fecfe647c7fae","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-15 09:03:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12af89dfd2ff954ecd61adc00a338b213f94c70b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 09:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f4cb4141303f7be071d1708ee2ea57db43cdc3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 10:00:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b7e3cc4b294186b092045e82ba368a : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3528ac632e444b683cfc70291d7d7c6 : SUCCESS in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/473d7ef3f2a04234b0b1e989c3738ac2 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed8794971dd49ffb445ab76279749bc : SUCCESS in 8m 56s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b065d42abead4a2191be82a01ef0d95e : SUCCESS in 42m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89a50d6fd6654e90a508eb9a0de9885f : SUCCESS in 49m 22s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b721850b8784cbfbf9caf1a075f3e46 : SUCCESS in 38m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/84012ae048934114b09e6614d931b0e8 : SUCCESS in 40m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/215e304c6cd24bdbaad526bf2193e534 : SUCCESS in 32m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d9549d7015bb46a3a994ead9ce709c34 : SUCCESS in 30m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/806c63d299ac4c03843365c8f7549c36 : SUCCESS in 39m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9188dfbe86634a88b2de450d479e9e61 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"73b6d1d3f4a3f7124e48686ec6f94ea6e8d8139f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 10:00:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bba92f30b0592ed14f0a3c31b09bdcee4569f53f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 10:02:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f44f1b81b5a4b91b5cb8edf9cc31c0b : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a1736f0709d7316ae9b7bf6fa9b488c7a342009","revisions":{"2a1736f0709d7316ae9b7bf6fa9b488c7a342009":{"kind":"REWORK","_number":1,"created":"2020-07-13 07:15:15.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/740622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/740622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/740622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/740622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/740622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/740622/1"}}},"commit":{"parents":[{"commit":"e049c5014a634e37dd3108b27de9e582ead499fb","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e049c5014a634e37dd3108b27de9e582ead499fb"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-07-13 06:02:48.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-07-13 07:15:15.000000000","tz":0},"subject":"Add a migration that deletes redundant delayed calls","message":"Add a migration that deletes redundant delayed calls\n\n* Due to the bug https://bugs.launchpad.net/mistral/+bug/1861988\n  some Mistral installations may have redundant delayed calls with\n  the empty \"key\" column. This patch adds a DB migration that cleans\n  them up.\n\nChange-Id: Ic46d3d711a03fbd6364809a0fddcbb48f3e28d2a\nCloses-Bug: #1887335\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a1736f0709d7316ae9b7bf6fa9b488c7a342009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a1736f0709d7316ae9b7bf6fa9b488c7a342009"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
