)]}'
{"id":"openstack%2Fmistral~741902","triplet_id":"openstack%2Fmistral~master~I8f5ee8ae9542d9a15fb0937b4a9de1db9915a3a7","project":"openstack/mistral","branch":"master","topic":"v8eval-py3-compatibility","hashtags":[],"change_id":"I8f5ee8ae9542d9a15fb0937b4a9de1db9915a3a7","subject":"Fix Python 3 compatibility issue in V8EvalEvaluator","status":"MERGED","created":"2020-07-20 09:40:11.000000000","updated":"2020-07-23 07:47:20.000000000","submitted":"2020-07-23 07:45:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741902-1595490356926-bfbdc47d","meta_rev_id":"483412eb04a5dab63fe339de2c476eb03a1be60d","_number":741902,"virtual_id_number":741902,"owner":{"_account_id":23383,"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","username":"endre.janos.kovacs"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 07:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-07-20 09:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2020-07-21 14:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-07-23 06:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-07-23 06:52:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-20 09:56:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 09:55:08.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-07-20 09:56:00.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-07-21 14:29:26.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-07-23 06:52:40.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-07-23 07:45:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c2f9fdca84c7d28876245b5b06d57c609b52df8","author":{"_account_id":23383,"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","username":"endre.janos.kovacs"},"date":"2020-07-20 09:40:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4e7b4c9228675e0a889729993128b62f5f9327f","author":{"_account_id":23383,"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","username":"endre.janos.kovacs"},"date":"2020-07-20 09:51:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"df7f92ea13d5cbcf88da8fad068de7dd9e02fb69","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-07-20 09:55:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"67a142993872833490fce646db4004626247cc68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 11:31:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9785c388f5453697aaaf0b02868bf7 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b6dcb118e124f498867bfb826eb209b : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a764946468d4cc997fa3ec50e556024 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126c45999b794e099f3aa216147095c8 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de843eecafd64e3fa15d9baace6b11c9 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c27d86093164eb089ec53f93030aca9 : SUCCESS in 9m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ecfc10af4469454188157faafa1896fc : SUCCESS in 38m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ec4b91c0edb44bab1aab323c50b1816 : SUCCESS in 39m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54fb67aad98f4dd592f77def60b12e30 : SUCCESS in 44m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/832257ec90fa4909b331e70c10b58404 : SUCCESS in 45m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4764fa0435d344ae8998c5b85d14afcc : SUCCESS in 38m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fe32de22531d433b90b7242072399917 : FAILURE in 31m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/933e8cf5c4f5428b85845d8b71dd71c7 : SUCCESS in 29m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/56a43b6b47f148208a49bd95a4a1a9c6 : SUCCESS in 1h 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da54b52aebe7ba86077607d80364e3ca1568690f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-21 06:52:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd00181997e88f3f63d6d20d0b9bb5f30fd421b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 08:17:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08a0d9b33c941b19f3dc4a08c750811 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73ea48dce0f8417a8ff5e0b0a8e1f4ec : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79e0ae280b544caba895f4775a43ae6 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d537d120de4474fbddaf214652cfe0d : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16dfe0588b2a4cfcbdf724c8c6634ac5 : SUCCESS in 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07807757b79e4558a479eafae6538ca0 : SUCCESS in 10m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b16ef664e9e84c5fa8fc2b39015a5724 : SUCCESS in 40m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b52de8fffd940bdb70ab6b9ce888fb0 : SUCCESS in 40m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d573d63bb3d4b82b5fb1aced34590a8 : SUCCESS in 39m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7e54f64b4fb54b43b68b55086336ae3a : SUCCESS in 39m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f3cab8f9d3724cb2a7a634a53a9cbe95 : SUCCESS in 37m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f4c35d59082c45218bfef28ffb63ae63 : FAILURE in 30m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/656a0901b3bf401d80053d4312fc2621 : SUCCESS in 28m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6262fedcc0b34a079cc1ad66925d0fd7 : SUCCESS in 1h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b127231807116cc5b09658c1392cc0c8fcd40239","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-07-21 14:29:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c81f21be5cf8599343a454157d8e8ca9f098c3f2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-07-21 14:29:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f8480d54a087dd760f82eec1b9fe121621088ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 16:45:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b20c5677694b2499197ed8402099e9 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5a0a9428e3e440ca1a4432c9926c8e2 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2361d19e8bd4283a0a000b1aeb45cfd : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229085da2a1c487f91aa32d906b055bd : SUCCESS in 9m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ba23c0a37a243bf8b6005ae1587e742 : SUCCESS in 11m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71fb68c2f9b846deaf56129bb8ef463a : SUCCESS in 11m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/04c27769db224c348a5ee96be1aecf4c : SUCCESS in 45m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c544a87c1abd41d39d62994c97fd64bf : SUCCESS in 44m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a4621e93f9d436bad73522d6faff6e2 : SUCCESS in 45m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fa07030ce0674b0fb23e7a9db81e9621 : SUCCESS in 41m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5c49a030b7e54fb3a32c312c808d7f09 : SUCCESS in 40m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3bc29f5b7ec04848a6ddce3c10acda99 : SUCCESS in 29m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/93b70c66a66d46e7857738eaa3464ddd : SUCCESS in 28m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/689504dca93049c78468b0fbc31a0c73 : SUCCESS in 1h 54m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13e9e7ef8015c814aac54708c6eac4284c57a801","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-23 06:52:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2dcd86443e52234da43a7825eecfaa2855c3e5dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 06:52:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"908c5012155f8fbb1c2f5746ad2a9f078e2b84bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 07:45:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"862ebd53f3616753350b6539fa5cb47fa4762463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 07:45:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2264e6952571488bb86c3db7e182d37b : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ed9934f44f473080995397d7edf713 : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a727a8082534f24bddae8658a7bd2ea : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5a801d9c9b44a1b744925415ebce40 : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08f09ba0ecdf4c989ac1d8c42f656852 : SUCCESS in 11m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1aa396a1bc7b4c809f7294324af72120 : SUCCESS in 33m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/603a106cf30944128c70bd5e223e9fd5 : SUCCESS in 31m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce4d218cb769441aa631c26596bfeeca : SUCCESS in 32m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b4bfbbf165344784a75ae3aa4544e161 : SUCCESS in 45m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc40f78223694b039fbaa9c0ea301d6f : SUCCESS in 43m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/32b1e2cc30d24bfb921411c51acc4ae1 : SUCCESS in 31m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/19ede3327b924b969838147b0ed31424 : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":2},{"id":"322b349eab4a325ef92a0f95a3b1e755959f1e09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 07:47:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492fee6477b042928060e8accbe1e0c9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0d865063b56c76f45d38bf99c722d2efff02a92","revisions":{"ee85ab6fd5194003c41ad896e25727898c702860":{"kind":"REWORK","_number":1,"created":"2020-07-20 09:40:11.000000000","uploader":{"_account_id":23383,"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","username":"endre.janos.kovacs"},"ref":"refs/changes/02/741902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/741902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/741902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/741902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/741902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/741902/1"}}},"commit":{"parents":[{"commit":"703261c6ae6eaf83516e726b570054b652614fa3","subject":"Merge \"Add a protection for list commands from parallel deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/703261c6ae6eaf83516e726b570054b652614fa3"}]}],"author":{"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","date":"2020-07-20 09:31:35.000000000","tz":120},"committer":{"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","date":"2020-07-20 09:31:35.000000000","tz":120},"subject":"Py3 vs V8EvalEvaluator: encode() is now not only unnecessary but actually breaks the action by converting the input str to bytes","message":"Py3 vs V8EvalEvaluator: encode() is now not only unnecessary but actually\nbreaks the action by converting the input str to bytes\n\nChange-Id: I8f5ee8ae9542d9a15fb0937b4a9de1db9915a3a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee85ab6fd5194003c41ad896e25727898c702860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee85ab6fd5194003c41ad896e25727898c702860"}]},"branch":"refs/heads/master"},"d0d865063b56c76f45d38bf99c722d2efff02a92":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-20 09:51:27.000000000","uploader":{"_account_id":23383,"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","username":"endre.janos.kovacs"},"ref":"refs/changes/02/741902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/741902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/741902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/741902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/741902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/741902/2"}}},"commit":{"parents":[{"commit":"703261c6ae6eaf83516e726b570054b652614fa3","subject":"Merge \"Add a protection for list commands from parallel deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/703261c6ae6eaf83516e726b570054b652614fa3"}]}],"author":{"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","date":"2020-07-20 09:31:35.000000000","tz":120},"committer":{"name":"Endre János Kovács","email":"endre.kovacs@nokia.com","date":"2020-07-20 09:51:20.000000000","tz":0},"subject":"Fix Python 3 compatibility issue in V8EvalEvaluator","message":"Fix Python 3 compatibility issue in V8EvalEvaluator\n\nstr.encode() is now not only unnecessary but it actually breaks the\naction by converting the input str to bytes. Since Python 2 support has\nbeen dropped, the call to encode() can simply be removed.\n\nChange-Id: I8f5ee8ae9542d9a15fb0937b4a9de1db9915a3a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0d865063b56c76f45d38bf99c722d2efff02a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0d865063b56c76f45d38bf99c722d2efff02a92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
