)]}'
{"id":"openstack%2Fmistral~742087","triplet_id":"openstack%2Fmistral~master~Icce8dfa6edb18973ca2a4eb594818ea9d504b49a","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Icce8dfa6edb18973ca2a4eb594818ea9d504b49a","subject":"Generalize caught exceptions on workflow delete","status":"MERGED","created":"2020-07-21 07:44:54.000000000","updated":"2020-07-21 11:32:05.000000000","submitted":"2020-07-21 11:30:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742087-1595331012113-bdb66bb5","meta_rev_id":"962eefe14f28997b7aa9b864370b6da769b39b76","_number":742087,"virtual_id_number":742087,"owner":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 11:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-07-21 08:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-21 10:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-21 10:32:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 08:59:56.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-07-21 10:32:27.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-07-21 11:30:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4d2ff778f4cf1f7d13fadc42a58b64fef3a117c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-07-21 07:44:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3316cf516f1c762c60cd5fd3a3b435ba65c5214a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-21 08:09:11.000000000","message":"Patch Set 1:\n\nI see in the Zuul dashboard that it\u0027s still failing.","accounts_in_message":[],"_revision_number":1},{"id":"70517ed1a81526263632b8c4dd829a3b7ccccc29","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-07-21 08:21:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b671a3b9f4bc7089e7498edd49ddb17199bbbdb7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-07-21 08:59:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7f1c6128c84a65a090b2f0de4c0a3c7a168c19f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:13:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebb2c285b5154ff997550774268392b4 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a30f94dcc2c49119887a717f1661ca8 : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef4ca7d09e84cbf8b5f983ab7c473e8 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b97b39e38245439b38b001306fea9c : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1ae489edc5f4f109666860e8b1e9770 : SUCCESS in 11m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d260b1bc7cf46e2a48faabdf499dd30 : SUCCESS in 11m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9233b3698ff046fbb7d23da59f82c9c5 : SUCCESS in 46m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/811fc89ba03041a5a3628ec4841e84fa : SUCCESS in 48m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c41608e448c541d4946268fa5241d4fa : SUCCESS in 49m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/93e4de876b9540a1b2ac2e3941a8953f : SUCCESS in 44m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ab7ad2f08be441e6803c90adf33f4185 : SUCCESS in 48m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0f6ad7da142d43f39f0ed86ce30b1e5f : SUCCESS in 29m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2a5ad9896f13408695cb7d40b461771a : SUCCESS in 30m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6504023c09534e2c99d29c1d05f679c8 : SUCCESS in 1h 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e11aabbd062ced880087f9e95290ffb132481881","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-07-21 10:32:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06a8fc2f88ac2b4a365043c7cbf77a58d9fed8ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:32:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c0b0efb3477d363d7bc87ee4eb6507e0c474a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:30:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00915a24bbcd4ff0acdbe5bc0d908203 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6408bb3a34b343a387a13e78b1f16056 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ecedf750b594a5196cd626c823b905a : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7078e20b32447e3b13a2f1e9324da51 : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f4a4aaaa7a94d46b768f4ec77ae09cb : SUCCESS in 8m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/14b84606c8bf441b9b4fb37e9d75342a : SUCCESS in 47m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23e7e51cad7e4a72b6cf42e7780da767 : SUCCESS in 46m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83e84e282e094ed78827269e9bcbaae5 : SUCCESS in 46m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a0c64669c10c4fbb9b009d0353a4b565 : SUCCESS in 47m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f5d6cb493eba439280979247e40972ea : SUCCESS in 49m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c733ba5c3f1a4d1d9a7ce628029d185a : SUCCESS in 32m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38399f6e3c0545cdb0fd198c55f820d3 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3c60964789c42a7f51976ac8b7fee157978b2c7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:30:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"962eefe14f28997b7aa9b864370b6da769b39b76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:32:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b407c49d177d422480dbdddaed79887c : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e600ce6fc91a4be90413ea126095a82a26b9477","revisions":{"4b6d167149db028c82a17f9741af8e752b71d894":{"kind":"REWORK","_number":1,"created":"2020-07-21 07:44:54.000000000","uploader":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"ref":"refs/changes/87/742087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/742087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/742087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/742087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/742087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/742087/1"}}},"commit":{"parents":[{"commit":"703261c6ae6eaf83516e726b570054b652614fa3","subject":"Merge \"Add a protection for list commands from parallel deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/703261c6ae6eaf83516e726b570054b652614fa3"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2020-07-21 07:36:48.000000000","tz":120},"committer":{"name":"Andras Kovi","email":"andras.1.kovi@nokia.com","date":"2020-07-21 07:41:22.000000000","tz":120},"subject":"Generalize caught exceptions on workflow delete","message":"Generalize caught exceptions on workflow delete\n\nOsloDB sometimes does not convert the underlying SQLAlchemy\nerrors into DBError. Sometimes it happens that the \"too deep\ncascade delete\" error manifests in sa.OperationalError instead\nof DBError. We generalize to catch both cases as the underlying\nissue is the same (just as the message of the error).\n\nChange-Id: Icce8dfa6edb18973ca2a4eb594818ea9d504b49a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b6d167149db028c82a17f9741af8e752b71d894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b6d167149db028c82a17f9741af8e752b71d894"}]},"branch":"refs/heads/master"},"2e600ce6fc91a4be90413ea126095a82a26b9477":{"kind":"REWORK","_number":2,"created":"2020-07-21 08:21:06.000000000","uploader":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"ref":"refs/changes/87/742087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/742087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/742087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/742087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/742087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/742087/2"}}},"commit":{"parents":[{"commit":"703261c6ae6eaf83516e726b570054b652614fa3","subject":"Merge \"Add a protection for list commands from parallel deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/703261c6ae6eaf83516e726b570054b652614fa3"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2020-07-21 07:36:48.000000000","tz":120},"committer":{"name":"Andras Kovi","email":"andras.1.kovi@nokia.com","date":"2020-07-21 08:20:57.000000000","tz":120},"subject":"Generalize caught exceptions on workflow delete","message":"Generalize caught exceptions on workflow delete\n\nOsloDB sometimes does not convert the underlying SQLAlchemy\nerrors into DBError. Sometimes it happens that the \"too deep\ncascade delete\" error manifests in sa.OperationalError instead\nof DBError. We generalize to catch both cases as the underlying\nissue is the same (just as the message of the error).\n\nChange-Id: Icce8dfa6edb18973ca2a4eb594818ea9d504b49a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2e600ce6fc91a4be90413ea126095a82a26b9477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2e600ce6fc91a4be90413ea126095a82a26b9477"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
