)]}'
{"id":"openstack%2Fmistral~744345","triplet_id":"openstack%2Fmistral~master~Idb053c8b1bfde9870eea80f66e5aef156422a668","project":"openstack/mistral","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Idb053c8b1bfde9870eea80f66e5aef156422a668","subject":"[goal] Migrate testing to ubuntu focal","status":"MERGED","created":"2020-08-02 00:22:22.000000000","updated":"2020-09-18 06:30:32.000000000","submitted":"2020-09-18 06:27:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744345-1600410446259-e889b14a","meta_rev_id":"37a4ea34c785f98f88f16e5b490b995a8f92be1f","_number":744345,"virtual_id_number":744345,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 06:27:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-17 14:46:52.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-09-18 05:30:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-09-17 14:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-09-18 05:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 14:47:55.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-18 05:30:15.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-09-18 06:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26d1a83e321161465e53bb5a5664a41a55494510","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 00:22:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1c97db8b79d9a364dd006e9e21c03f5b2e204c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 02:11:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afbcd48e59654487936b0a92f292ff74 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74c2c353a375417eb8bb85acb7aea1f7 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99d6406889aa4e78a2c0d502454c0019 : RETRY_LIMIT in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9ebc988bde48f2a61542122d925029 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b20248e7c203414aae7eaf44baf3eaa8 : RETRY_LIMIT in 4m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cb5be7d6448a4388aee488a8c379bb62 : SUCCESS in 38m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/708d3d3af2804253b43bea95f9cb35a8 : SUCCESS in 46m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3899f947148e4ed596bd2c5afe1adcb7 : SUCCESS in 45m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c6479b84d5524369b4d642803d8f1fca : SUCCESS in 41m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/213af90995b048e7ab41f23c32fac2f8 : SUCCESS in 49m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9ee4cbbd70d45e8964d777abde1f16d : RETRY_LIMIT in 4m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/145c04d6802b49d4b3227d5cb8f3d8da : RETRY_LIMIT in 3m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/04bc3527ffdf4ae8aa1c5fc50ed8f2d9 : SUCCESS in 1h 41m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3394f99c439e52baa44e31425f5d72875abfaa94","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 23:04:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"339e26c869eaddafcb3ee307ad810582978c5d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 00:39:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e894edba819044be9cb7cb92a2f6dd7a : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c45e777c6784969a772ae69c1386ca4 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c21e1a4599c24afa99dd87e4c6e50126 : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0169689507f94ca0ac5392d82739ea82 : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c257bedc796465682372e4a508a69bc : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2afbe48d8d884223abc9aea45f376532 : SUCCESS in 8m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e44e0073257744f7ab78498a3e666ce8 : SUCCESS in 1h 00m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61abb7a7a5c34eff87a33cd4c4550e46 : SUCCESS in 38m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f230a3e82f194494b6329dab89a40f8a : SUCCESS in 51m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/412613f9188a418d91e9cd924370c5ad : SUCCESS in 57m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9d1f81fc58564dbcbe326675c0b91ed9 : SUCCESS in 31m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/225e4eabbde74e2ea11daee0668a93d1 : SUCCESS in 30m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8d4a23e582654652ac85121c3f462971 : SUCCESS in 29m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/27599ea6b0034c5d88c08db45bcb1978 : SUCCESS in 1h 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"618ba3c0e125c1879e2c433bb23eed9d3a564c61","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-05 01:41:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4ce5aaea6e1b2941bbc2054fd95bbe4b1f818c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 03:31:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6546850100094f44bf432570cd96f4a3 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14feb52a23984d88805cefb0bf35f1db : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cd66f06b9f44b0296c89d7566a47cd8 : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09dc88a11064856a04ef34484cc560d : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e5c931fc9ee49c886f74d680d60f1c6 : FAILURE in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25f9602d83e4f8490ac8f35f479d145 : SUCCESS in 12m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/560a222e399b48bd9334bf8933217cb1 : SUCCESS in 41m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40e442dbfeb447f194cacdf9c8532e82 : SUCCESS in 42m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10494c31969e4c659f6ca93832d540d9 : SUCCESS in 43m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6e10a245a89640c48bf8a6925b9f0a26 : SUCCESS in 42m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e4befd751b794e6dac8b4369da87aba1 : SUCCESS in 41m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7ad463a1af9d432baa7a25aeefddb982 : SUCCESS in 29m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4c02b6fae89a4b6ba7028ee9bbc2c865 : SUCCESS in 28m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b46e435162c142558bb680376462c166 : SUCCESS in 1h 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8bdb23e27622b416da8d038418435861a5ffb26","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-16 23:18:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9f310c1a995d2114e4b4fae329cd18f0dc0fdfa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 00:42:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0605281dcc0e42cab25b706b471a5184 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f55c78a10b748988a0388d35ec8def3 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8bcc2275ffb4f6ba12e0a4152b02e49 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2852a71cf5094eafa16ee82928be6315 : SUCCESS in 18m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a55758b699b4012b65e8c66c63b3554 : FAILURE in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf2b4148f3ce443b99a1dc4e5a292710 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a03a12ed3e5c470e98eb64202220a478 : SUCCESS in 10m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3770adb055604c048a4f632a525cfa90 : SUCCESS in 45m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7e330aeb5849c6be3acbb5cc97bd61 : SUCCESS in 45m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c4ac5584dc748cebdb39b5b84a0615d : SUCCESS in 48m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/469e47166d6d42788c869f5875a1e280 : SUCCESS in 42m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc1ca7a1e15044a589da630760803013 : SUCCESS in 52m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ed09736e22b64af69db5e76471c19c28 : SUCCESS in 30m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e39b1cdd3aaf44b4b93643b3534d7baf : SUCCESS in 29m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/99b29cc4935442278f9db2c06f921216 : SUCCESS in 1h 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6bc9bb7d8b6a8be499853e00247b13516c3673eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-08 21:55:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3ac03a3d269c4055f6af0707477791a90e98f685","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-08 21:56:04.000000000","message":"Topic changed from bug/1885825 to migrate-to-focal","accounts_in_message":[],"_revision_number":5},{"id":"da3fb9330c7c1a5426e6f88cd3d9e0096b97d262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:46:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0dbf4f448f4bf293a4d23c15199205 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b95d2e5b54a547eabe87cb7d8a9c79d4 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dca7031b3ea04c4a94f19e8070d86830 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692c037a762d4d22a6ed5c81d5a7c930 : SUCCESS in 15m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/79825bd339bf4c658f332b4852e7769d : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d350ec5ee79a4c72b89ef0aea9102709 : FAILURE in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44980733d41b4e55832336180ee589dc : SUCCESS in 9m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3cf4b041e5a7434b840953b4d27a0289 : SUCCESS in 55m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2169e5ab2f714d8988a7bb2ac8a4f714 : SUCCESS in 55m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ddef2a09bc34799977148767e8dd9ff : SUCCESS in 51m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dec22869143c4a7ab2c4441db4e4af9e : SUCCESS in 49m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/47d19a78e65e47939a6791ca96a600de : SUCCESS in 53m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/13457817751b4843875d2d296f7af83e : FAILURE in 29m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e8a8d64f7dbd49f0b4ae24208913295a : SUCCESS in 30m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1adb6eea18a14634833e733d5b8a88f3 : SUCCESS in 1h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d151c3c28400ef96918ddd393963b7ab03c3e796","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-16 20:38:25.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ea71c1beee8f77d1ff1df46f2568d87013fe7b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:17:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a7a878e774449c8beee9f333721b830 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc72ba9d24c140edb68ad449a8a692e1 : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d368088e74441bfa58f23745f37ff82 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d24b956d16e4d67b45026e64fead92b : SUCCESS in 17m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54f96c614afc447b8f8cb7eda8e5927f : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05ea9e7411d04c70a2d04dfc4aa9c6c9 : FAILURE in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/898e7ecaaec9451190096321308a7e8f : SUCCESS in 9m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cbe3eae44d8347028960b470af1fdb14 : SUCCESS in 48m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57dd9ff9bdbf4451b80f7c67e3729f6f : SUCCESS in 59m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce2af58afa514fbb8b143b9bc9be5b27 : SUCCESS in 48m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3c01086e1d584c4486cff725b3fe7ea7 : SUCCESS in 49m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/558ed3e7de7f41f18b6eca4117996a9d : SUCCESS in 56m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/45ce89e9cf7244bab7a0329d2d632b0c : SUCCESS in 29m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1557ca1de3454cb48009a1904e4ab001 : SUCCESS in 28m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/eb4a16fdde7c46ebaa5aa2f77f8cf49e : SUCCESS in 1h 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b08c382d150b7a89b3b96a333ccf43a9d44e2625","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-16 22:37:06.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"70cd0ee8246a173c6ed50350acfb321f437b02d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 00:17:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41bee4ead3f54fcdbc8065ceb68207eb : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c674fe547a4b8db6ae4a6b2c78a309 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4a23cb8a0c4351ac5ebad9aa351a2a : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d590141dfc4e8a9cd50da78ec7e9a4 : SUCCESS in 18m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/62aa5c93e1fa4b46853f778ab7f3b1b7 : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de7af2515d4649af8e462d16ee41a277 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6828c96114946f58a55b92ba926f339 : SUCCESS in 12m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/42dfa755172a475cb987aebfbe88e368 : SUCCESS in 46m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3a05a8838a44cb889c4190ada005677 : SUCCESS in 45m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a49a5f4b40043139cd4f8fdb7623a36 : SUCCESS in 43m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/373e114eb92a4b48aa92bdbde330ea24 : SUCCESS in 43m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ceaa6ecf57ba42a08ddad99fbe55292e : SUCCESS in 43m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b2026647a5da4081ad06ed6993584825 : FAILURE in 34m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/11db72b4a4364752877f5a078efe9760 : SUCCESS in 28m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/706d7ab2aef947adae0213e20549e984 : SUCCESS in 1h 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d6eebf3b509b652a0d27a7a445e062179d3d5fec","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-17 12:31:54.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"6f3a101b53fada980bcd5d2d60f0a7dc7791b2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:43:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d225ccfcfb3a4f85ba648ca8ad7a1aa0 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0242281f1824d7d814776ea4bd44b76 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5724a28a7ec94247b4f4fc4afe53821b : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7770c8ff207c444185407a8fac2807c4 : SUCCESS in 15m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c72342da5c945ea843bf87d442320d7 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2474faedbf804b69b4ab9e32db46d633 : SUCCESS in 11m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb99735127be4988934ae6f35c624fc4 : SUCCESS in 9m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9d87bbd0a06249dd8f39c8c1a131919c : SUCCESS in 55m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56b2a0064818472ca8f28bc035346b59 : SUCCESS in 49m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/424110e556a944419d8bc17124a849e6 : SUCCESS in 54m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/252daed62d234d77ae208b7126a878d3 : SUCCESS in 50m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cbfb3b3558294a1db77d40a34982d88b : SUCCESS in 50m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b39ec7346cf04c5596e7e90942b9481d : SUCCESS in 32m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8fa9d9c8dc0b45d48c1c595265c70831 : SUCCESS in 28m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3f1d2084b23d427a9254992397fa04a4 : SUCCESS in 1h 57m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f2fb7edb138457fa20c193d8f8c1d813a0b04db3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-17 14:46:15.000000000","message":"Patch Set 7:\n\ngreat, let me remove the depends-on to merge this and fix the gate","accounts_in_message":[],"_revision_number":7},{"id":"39a6367fe2297a186404be622f4ea89ba87d9763","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-17 14:46:29.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"f058294e88b878e23c6f9e052202eaaff45ec548","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-17 14:46:52.000000000","message":"Patch Set 8:\n\n@Eyal, this is ready to merge now","accounts_in_message":[],"_revision_number":8},{"id":"67c6c2c07cb95c58e817c5a52bc31ec25abb57c6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-17 14:47:55.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"43937b48a395eef2a8af7230b7a1962e6b3cabc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 16:32:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6501a7034741a19f5daf9f1d971247 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6806d786e3174872963d88fb8ec0a9ef : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ed3070dfdad4c1fae5f8851be2f67a6 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b66f1f0f078446bbbc239b070605b86e : SUCCESS in 32m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc4068e2bafc42529b7828c89cc65b75 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b83848c8f4ec4afdaac9ae14037ee950 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a46a3c6d98944f19a9037c0d63001054 : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a3a78d042f9d4cc88f4b624723bef769 : SUCCESS in 49m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4c09326e69a451e967047ff794dad95 : SUCCESS in 48m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb2e66c6bf924417ad4add9dafb2ceb1 : SUCCESS in 49m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9054b7c374f344e18fc60743d47cf863 : SUCCESS in 47m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f556f97eed134130b3a2efcbe53bee88 : SUCCESS in 52m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e676d95bc99c4c50bc8a89219878d0f8 : SUCCESS in 29m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/98ae9e09c2ad4c41848e9e5a1605e95c : SUCCESS in 29m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/89020847f58246489cb5c175e62b20bb : SUCCESS in 1h 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a94a9a4378228df00da0491d539bdc8d429a8b36","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-09-18 05:30:15.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"46abbac23d0b861b26ee89ae852fa6e43a1be52a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 05:30:27.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8072801b20317383762e9ec5929ec6a94213fabb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:27:25.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84a36808c3b434dab4db1dec1c87fbd : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aacb7cab41b4650a58a95d2c1685f1f : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aa96ecd2211465c87bfdd5d592f9209 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b795251934a4d648128f0c99dafab8a : SUCCESS in 16m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f594b9c4a87643cbb09f935059088088 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beef4e489712471cb0046813c7851f0a : SUCCESS in 9m 13s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/53ac08a7131d47a28d675d1480a510ee : SUCCESS in 48m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f837b2d7d35c42959df2dec0b0097e6b : SUCCESS in 41m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a18fefc1f2974cb18ac998698c676a19 : SUCCESS in 47m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/15ca346ea648450ca4f9ff2bde9b355a : SUCCESS in 51m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b24db78fcc87429fa722312b5af24a9f : SUCCESS in 40m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/781494bbfbb34adcaa8481339e411939 : SUCCESS in 30m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7c0cf182d22b44439186dcc5e23fc497 : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":8},{"id":"6866dde4b1f45e8797d7febf1d0405e4467be79f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:27:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"37a4ea34c785f98f88f16e5b490b995a8f92be1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:30:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445f84a115334fa6af8a74684c1e97b8 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b55cc013c1611fda2e589435f8ba28a9225e7827","revisions":{"83b664fbd6d0283c3383d2053e2ee61befcc0ad3":{"kind":"REWORK","_number":1,"created":"2020-08-02 00:22:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/1"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional tests jobs\n- docs jobs\nDepends-On: https://review.opendev.org/#/c/734700/\nStory: #2007865\nTask: #40196\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83b664fbd6d0283c3383d2053e2ee61befcc0ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83b664fbd6d0283c3383d2053e2ee61befcc0ad3"}]},"branch":"refs/heads/master"},"09c48772c6a8d47790305bf0bed7b1455d0a8f4b":{"kind":"REWORK","_number":2,"created":"2020-08-04 23:04:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/2"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 23:04:48.000000000","tz":-300},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/09c48772c6a8d47790305bf0bed7b1455d0a8f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/09c48772c6a8d47790305bf0bed7b1455d0a8f4b"}]},"branch":"refs/heads/master"},"299a0d94140c72389a8b1533667e2e344e9aa407":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-05 01:41:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/3"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-05 01:41:28.000000000","tz":0},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/299a0d94140c72389a8b1533667e2e344e9aa407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/299a0d94140c72389a8b1533667e2e344e9aa407"}]},"branch":"refs/heads/master"},"fff63f81f871687173ade54d86f28c6ae631696c":{"kind":"REWORK","_number":4,"created":"2020-08-16 23:18:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/4"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-16 23:17:25.000000000","tz":-300},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fff63f81f871687173ade54d86f28c6ae631696c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fff63f81f871687173ade54d86f28c6ae631696c"}]},"branch":"refs/heads/master"},"5de42b860ad0aeb79a4cc1852f423a142d347ac1":{"kind":"REWORK","_number":5,"created":"2020-09-08 21:55:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/5"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-08 21:55:25.000000000","tz":-300},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5de42b860ad0aeb79a4cc1852f423a142d347ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5de42b860ad0aeb79a4cc1852f423a142d347ac1"}]},"branch":"refs/heads/master"},"6897c7f27cdcf2cd1d85188113d2489d008dac4a":{"kind":"REWORK","_number":6,"created":"2020-09-16 20:38:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/6"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-16 20:38:23.000000000","tz":0},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6897c7f27cdcf2cd1d85188113d2489d008dac4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6897c7f27cdcf2cd1d85188113d2489d008dac4a"}]},"branch":"refs/heads/master"},"b46e94d067ee91db66e21ef1a8151da85ae7bdb2":{"kind":"REWORK","_number":7,"created":"2020-09-16 22:37:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/7"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-16 22:37:04.000000000","tz":0},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b46e94d067ee91db66e21ef1a8151da85ae7bdb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b46e94d067ee91db66e21ef1a8151da85ae7bdb2"}]},"branch":"refs/heads/master"},"b55cc013c1611fda2e589435f8ba28a9225e7827":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-09-17 14:46:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/744345/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/744345/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/744345/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/744345/8"}}},"commit":{"parents":[{"commit":"598d214de8e6d46bb3758ec834c02e2d3ab663ed","subject":"Merge \"Fix Python 3 compatibility issue in V8EvalEvaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/598d214de8e6d46bb3758ec834c02e2d3ab663ed"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:20:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-17 14:46:27.000000000","tz":0},"subject":"[goal] Migrate testing to ubuntu focal","message":"[goal] Migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1885825\nFix user creation with GRANT in MySQL 8.0(Ubuntu Focal)\nUbuntu Focal (20.04) has mysql 8.0 and with mysql 8.0 there\nis no implicit user creation with GRANT. We need to\ncreate the user first before using GRANT command.\n\nStory: #2007865\nTask: #40196\n\nCloses-Bug: #1885825\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal\u003e\n\nChange-Id: Idb053c8b1bfde9870eea80f66e5aef156422a668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b55cc013c1611fda2e589435f8ba28a9225e7827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b55cc013c1611fda2e589435f8ba28a9225e7827"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
