)]}'
{"id":"openstack%2Fmistral~752773","triplet_id":"openstack%2Fmistral~master~I093f9886e74a808cedca0ee6610d51e91c88b1e8","project":"openstack/mistral","branch":"master","topic":"remove-translation-babel","hashtags":[],"change_id":"I093f9886e74a808cedca0ee6610d51e91c88b1e8","subject":"Remove translation sections from setup.cfg","status":"MERGED","created":"2020-09-19 01:16:26.000000000","updated":"2020-09-21 08:20:32.000000000","submitted":"2020-09-21 08:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752773-1600676275288-193fd9bf","meta_rev_id":"6f34babf9e91ce84766b22983569c9158743dab9","_number":752773,"virtual_id_number":752773,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 08:17:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-09-21 07:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-09-21 05:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-21 06:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-21 06:41:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 05:11:54.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-09-21 06:41:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-21 07:09:58.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-09-21 08:17:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"129eb7f234b6022c45b7b298c42ab38b2072c0eb","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-19 01:16:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8835806bf7c5c3270c521818570b43315038d754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 03:09:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/004c49debce94351b5ee5e88b8daf83c : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/743edc6342c141978819c2162f27e8be : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfadbebc64fb4afa8b6bf4342046fdd9 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb44478a1484cc3b311676a468c81a2 : SUCCESS in 14m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f3e4b129864f4bc0a018fa598ae87e3b : SUCCESS in 2m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26f4de39fc724d678bc347583a014643 : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7849b210660d4f1385bdea407647331b : SUCCESS in 8m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/eef61beeba8045b6ba832abf5ba366e7 : SUCCESS in 43m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d2907b84024569bb5396abc7ec1a96 : SUCCESS in 49m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f43d4d3ef2f448ba1d1da125990300f : SUCCESS in 46m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d3687a9105ff453c82e8511458aab740 : SUCCESS in 48m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/88045e0e24bd4005b30ea5c0cd2155e8 : SUCCESS in 50m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/64b983980638489283e6aefa6e62f58f : SUCCESS in 28m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/761f4210468c4628903711f0ca821ff6 : SUCCESS in 26m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9f335510a61e47a3bad4b9da598dd172 : SUCCESS in 1h 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffa7934586c574c3531ba3dfd0b196f16a3a45c1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-09-21 05:11:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"142ac43bda6c82dfec5d47bfe9429f157402593d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-21 06:41:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e345c3ceb3e5dcd50899526a9baf0fb626e85f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 06:42:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c71997946f40da34a8882f8994dbd3113eda59b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-09-21 07:09:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c13a64c73a931df72ecbfe2f07786bcbfb6b444c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:17:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b78284ce8c7cc362d390115968b3416f97a73fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:17:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648fd39c36894a1f91e846741613d0a3 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d31975007c2d4d3682bf29cf7a9a775d : SUCCESS in 9m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6195738c5cfc420a96e92de62c81eb2f : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd02fed69d04fe3af31f0e6dd9157bc : SUCCESS in 17m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/169e910b20c94790808994e986c2dc38 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d231350243c741db8d797f80333a2232 : SUCCESS in 9m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ae6e37899d5345aaae623c994b175b5f : SUCCESS in 48m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d02284e4bf1444580efa7a691efa247 : SUCCESS in 38m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17c13cf2cd8e4eb8b335c741f81c41ff : SUCCESS in 38m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/debccfd77f0f404d965f3045ac5c44be : SUCCESS in 43m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e1cfc4631ea54f49bc7877fc705749e7 : SUCCESS in 46m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b6ecfb1bd046417d894e3f105b9b8dfa : SUCCESS in 29m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/14d7a418a2aa4638878b20053f723699 : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b08c4ecb709df0a57c477ebbcf76d3ad7553758f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:20:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d2a4606c9446b28e60bf0e80f29328 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac84e6d8eea60015f99b6ba2a5db5a2217daf1d0","revisions":{"ac84e6d8eea60015f99b6ba2a5db5a2217daf1d0":{"kind":"REWORK","_number":1,"created":"2020-09-19 01:16:26.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/73/752773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/752773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/752773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/752773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/752773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/752773/1"}}},"commit":{"parents":[{"commit":"a8fdf9662e8b38009c221cabe5689e68f3060f16","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8fdf9662e8b38009c221cabe5689e68f3060f16"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-19 01:15:14.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-19 01:16:20.000000000","tz":480},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore,\nBabel can generate translation files without them.\n\nChange-Id: I093f9886e74a808cedca0ee6610d51e91c88b1e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ac84e6d8eea60015f99b6ba2a5db5a2217daf1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ac84e6d8eea60015f99b6ba2a5db5a2217daf1d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
