)]}'
{"id":"openstack%2Fmistral~752774","triplet_id":"openstack%2Fmistral~master~Iefdc99c338c7aaea18d535426c4676dbedb44f32","project":"openstack/mistral","branch":"master","topic":"remove-six","hashtags":[],"change_id":"Iefdc99c338c7aaea18d535426c4676dbedb44f32","subject":"Remove usage of six.add_metaclass","status":"MERGED","created":"2020-09-19 01:28:00.000000000","updated":"2020-09-21 18:00:46.000000000","submitted":"2020-09-21 17:58:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752774-1600711106052-7e3b2230","meta_rev_id":"e401434a71e887c5272e1df8e14a5e28d0579fb3","_number":752774,"virtual_id_number":752774,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-09-21 14:19:11.000000000","_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 17:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-09-21 05:47:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":2,"date":"2020-09-21 05:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-21 06:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-21 06:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 05:11:25.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-09-21 05:47:56.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-09-21 06:41:18.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-21 17:58:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db2badc506f6455fe5be6d5a3724219ae2dab8aa","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-19 01:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbc518aaece936f3b6cfedd704a5b6a392004ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 03:34:45.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fa4b3e94f7647199ffbbe3bc0699037 : FAILURE in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbdb2002652d45488f921209df9f90db : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b737c7a59514c039c89dd87b5f084fe : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d87fc15fa1d42929f230729d102a776 : SUCCESS in 22m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37a426e4131441f3be4f74fd641635bd : SUCCESS in 13m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/626d8a0f153948d8adc46aff66854de1 : SUCCESS in 12m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/eafe9b78abc04013a0c6131dad29925e : SUCCESS in 1h 19m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c9da09cbb724d55a61f437403cef78b : SUCCESS in 1h 18m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e0bc99be6b449609ce05e814e914fed : SUCCESS in 42m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2e5965ffc6f942ff9fc89f9e770edea3 : SUCCESS in 1h 19m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/89abeebba82745ff9f83f2b8fe8934d3 : SUCCESS in 42m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/27e3c7fa2e62407f98a652f60c2b23fb : FAILURE in 29m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c691ed7336f94803884853ecb21967bc : FAILURE in 28m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d24b5bc97c02443cbc5df5f294ba2056 : TIMED_OUT in 2h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49f631a70ab3389c737a129d8f6fd1979977f5ef","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-19 03:38:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1d950a66f9742a7af0573a625ff25470fcbc5cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 05:33:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c130c37f55408c92a399434d7251d7 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d6ae0d967f9420dbdb27c56b85a5bc1 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b301b370a5374d719a2efda270e825c7 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ea2bb65b1e458d913763c0e096a8e3 : SUCCESS in 15m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c468b8bdc8984c6e8275f0bd71e1b385 : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bac139acb9ad4e59a411db949d2405b0 : SUCCESS in 8m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/397b4594b2774fa7a77120fb1b6d7e8a : SUCCESS in 39m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b23d67acee084360bb7b7e11d4133253 : SUCCESS in 42m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d8d8c02ed5d4d2f8cd10051beba955c : SUCCESS in 44m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8cd3cc661916411a99babdf7eed7f11f : SUCCESS in 48m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2f140841a34543f896f27c03b3ad53bc : SUCCESS in 45m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/58d16849479c43668c80464976b8d1ec : SUCCESS in 28m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d287b53bac7c4c30904e24ff8b16ede3 : SUCCESS in 26m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ecbb866dc17e473cba02e212b27b33b5 : FAILURE in 1h 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65802f3dd82c0d18a511122913a6128cdc5aad2c","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-20 12:43:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6886607928cf320ae0c75eee85083c8e7d88f7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 14:47:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f426fb89a7354f428588d2fd5c80f7ad : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c96582dc35743ca9e152f0da795d3da : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c1fd04e783a4b8d94871a1cbd310b54 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefdd6448ce64f088bba452c80d80198 : SUCCESS in 14m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ff42459ab0d422cb398760189b6ce08 : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d30109415784d9880a46063e3ab8314 : SUCCESS in 9m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4e72de5866ae4fd5a2c13ded9e7f8512 : SUCCESS in 46m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6112f4272b634d7ab1de75bf620de4fd : SUCCESS in 47m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/569dbfa6d2414da6b82c0e4c1fb44ff2 : SUCCESS in 50m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8c5bcf80bb45446bbae5f995060a6eac : SUCCESS in 50m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7eb1bc29c4364656ab6aee53d1eabec4 : SUCCESS in 45m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6fb13b6080a04de5bdef2dcd3ede0490 : SUCCESS in 29m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bbf1920ec4db4bc3b481e15b8df57326 : SUCCESS in 28m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/22754059431b441ca7ebdfb1710dad8b : FAILURE in 1h 58m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88c7b0c76b6f91189f5ad3ccacbc18e5ef2a118e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-09-21 05:11:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5342d9960219f6824126e0913df0e4bf1c561cf","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-09-21 05:47:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a3239c0c97e6a9107be4e7c0ca0077e9271eb786","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-21 06:41:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc669b274f6c9369ebed5ceb952d9dc4b02f2cea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 06:41:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3b2034fd498d8ce14b8897b55fd1b1930ed13ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:51:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b9466c7866423db14221ade6c38f42 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3833586a0a14af980cda1ec9e7c9f31 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd99c7d99c74beb92aaf0cd3ef9b4be : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b05e08615554035a1f067759784a8bc : SUCCESS in 16m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27221660d902442ab483c584c97f4d82 : SUCCESS in 10m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6b3d6380287b4b189956e18c8b28f6c7 : SUCCESS in 52m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acdb87ff43634f1ba6358247881259ee : SUCCESS in 50m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7574a529f744f9eb1280bd9bb5a34bf : SUCCESS in 1h 04m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/94815fa9443947e496f527ad73b7710e : SUCCESS in 48m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4d56e1921b8448828d36163cb10ee1f4 : SUCCESS in 38m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e339a313da0a4f0ead93e04315bb87f4 : FAILURE in 30m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3c77baa75f094663898a825412bed559 : SUCCESS in 28m 25s","accounts_in_message":[],"_revision_number":2},{"id":"4804e042a16637085830a3eb8d32654148e0c6e3","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-21 07:56:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cf516fdd09c8769a8356c94aadc42c2a7c349ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:30:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674e22d8d5e04ad1bb1d1d90e84e5683 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f45fae05933460da6774590c50ef8d7 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abd98241d7934bf3bea99df35e59059f : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d247e1be3347ad81c3b15151713e94 : SUCCESS in 20m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5088751b5104f3a828a1d238d7e8d9f : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01a1e075692d4ef2bec57a291a488cff : SUCCESS in 9m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f044186540574170a08f710982e7008c : SUCCESS in 52m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5888e653b1047c08bd8556d6a9cda4c : SUCCESS in 56m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c83e0620e3cc4670bbcc041f1bcba144 : SUCCESS in 39m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c24273405965411babeeedd8d79d0c60 : SUCCESS in 46m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5839c4df7a7944029762f3f0ff1924c8 : SUCCESS in 47m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2c480f0244d54f0b8a74872b8a0d028d : SUCCESS in 30m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c296544a2484e87b7f097317004f72b : SUCCESS in 27m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5da608709c1747748006bee80cf2273c : SUCCESS in 1h 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d3829b3ca6902f3e6c27181f9bfebec45bc3908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:30:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63f9f2953272ba24e8cdff5ada792e547df57b79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:26:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76769e44ccc64009ab1ad98cefb0bd60 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b02c71c1785441c3b606e0e04f700de3 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86707d9c017249c98d4b65e8365a9186 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9869ea364ca44f6e9848d1d7260e7b92 : SUCCESS in 15m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/250112043af341daa44cccf5e1a04967 : SUCCESS in 9m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/81dd7f7a8ff64f05be45a0ae36c08069 : SUCCESS in 51m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/072822fc5bdd4e2e82069f44001db6f3 : SUCCESS in 44m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eed804e28d14adba11d1d05500f816d : SUCCESS in 39m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9b85c58ba03549f494c2cc998b5d6a0b : FAILURE in 37m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e373d8aecec04f76891c686478c455ea : SUCCESS in 45m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7f53e0bd5a4e425da2a399dd57b4e8e7 : SUCCESS in 33m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d33972e80185427eb2f084b8c6ac4149 : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8ac42acf573e73051fdd1319f703c66e6e5f6737","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-21 11:49:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ef5fdf275b9ccf6eb3a9682f791e9cd5430f4a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 13:30:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4066dc7e74c4a5da5c10e394aa67356 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4fbdc67eba94b47a662960e31f1af80 : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7df3ad95ca474bd79304fa42183e96f5 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e644326918a4d39845d1662085289eb : SUCCESS in 13m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b4029001f704ce7995f60896836b61c : SUCCESS in 9m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/911217a521084138aa1933bc12e9c53f : SUCCESS in 9m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4306cfbc4ae542fd810b21814d2fefce : SUCCESS in 46m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a58b205036554a0fb65828f50e01d167 : SUCCESS in 55m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f72f08fc18034daebeaa2f4e6d333225 : SUCCESS in 54m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/87fa9cac1f9242cb8d7698683f6d12c4 : SUCCESS in 45m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9aae63e7136c490b9349d7a147b2774f : SUCCESS in 51m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/791821159d8f412983584023b55b7209 : FAILURE in 33m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/324a3d407b4f46828a1ed207d94c160a : SUCCESS in 28m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d5e54044f4104e1cbe5c6dcbe4f04e33 : FAILURE in 1h 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13c0cb6b135539e87b1cdf3dc61af583d574b096","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-21 14:19:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"34bc642342bfd4653e6205870ab89314ab3dd19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:01:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450ced1026c3405590b1cdebec94737a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c97cd07542ed4496817f0c0b7c5f4cdf : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3fb7bf3fe254add9913bda59891e015 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31ae3bfb515401f96ccf9720df49567 : SUCCESS in 17m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ed21bab25fd4956a416fddf33b0991a : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a4e87315c4e4944953501b06138f859 : SUCCESS in 11m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/20cfedffd32741b2807e6a35a218a91e : SUCCESS in 50m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64590dd8e7ec4e2da4187657fe766a66 : SUCCESS in 39m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebee70e035f34687a44ab1f150929184 : SUCCESS in 42m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/61c9f914ae7f49a1be947cf35d93bbb8 : SUCCESS in 45m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e26025ca5610454692b64d364d12fe6e : SUCCESS in 42m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d905865b805a4520b38036a71f0d071b : SUCCESS in 29m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6c905b852cb84c38b27639d5c4b41598 : SUCCESS in 28m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9e24d26e549347f395c3893fd82fb71a : TIMED_OUT in 2h 05m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2683a16b44722844a73bb6ae95ad2a0fdab863b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:01:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96c16a30435a3217c6d0639b7efe34c50992dde4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:58:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72e1801fbe7437bbf4aaaf975cd4822 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cd9b0684b694d1681b21508343805d6 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a204e3222ff94c29b815f2b6d6c6b37e : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd42f49bb6c48e6a17574bf504a280c : SUCCESS in 14m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/363b71fd541a4654a84785e19ef1dbb5 : SUCCESS in 9m 32s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e0b58858533849d3b29262d39f7830ff : SUCCESS in 44m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74e6a1e82abe4730813970d1b07337e8 : SUCCESS in 51m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f24a72901a84d9f88f3f40126abf944 : SUCCESS in 45m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bebe16a1c9df47619f543143d647922b : SUCCESS in 43m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bb39045fd3d04ab0a388e14317196d19 : SUCCESS in 43m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/edd8f8a66dd849d6af7797cb7866a7ae : SUCCESS in 30m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b1cea7647089405990e2b78a8e6ec9b3 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0496a8e03879504ffd3654deef5d6737cefdc630","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:58:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cdce6b83e94c953d2198d864e8c656fb8963b3e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 18:00:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a674ebd5ba60423a8edf315e145e860f : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aba14934e7c5197ef7c045f3517c942c0e3bf98c","revisions":{"3328a13c52e76bf552550c7e3ae5c95767f81d45":{"kind":"REWORK","_number":1,"created":"2020-09-19 01:28:00.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/74/752774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/752774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/752774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/752774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/752774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/752774/1"}}},"commit":{"parents":[{"commit":"a8fdf9662e8b38009c221cabe5689e68f3060f16","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8fdf9662e8b38009c221cabe5689e68f3060f16"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-19 01:26:47.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-19 01:27:01.000000000","tz":480},"subject":"Remove usage of six.add_metaclass","message":"Remove usage of six.add_metaclass\n\nWith python 3.x, classes can use the metaclass\u003d logic\nto not require usage of the six library.\n\nSubsequent patches will replace other six usages.\n\nChange-Id: Iefdc99c338c7aaea18d535426c4676dbedb44f32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3328a13c52e76bf552550c7e3ae5c95767f81d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3328a13c52e76bf552550c7e3ae5c95767f81d45"}]},"branch":"refs/heads/master"},"aba14934e7c5197ef7c045f3517c942c0e3bf98c":{"kind":"REWORK","_number":2,"created":"2020-09-19 03:38:06.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/74/752774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/752774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/752774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/752774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/752774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/752774/2"}}},"commit":{"parents":[{"commit":"a8fdf9662e8b38009c221cabe5689e68f3060f16","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8fdf9662e8b38009c221cabe5689e68f3060f16"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-19 01:26:47.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-19 03:37:24.000000000","tz":480},"subject":"Remove usage of six.add_metaclass","message":"Remove usage of six.add_metaclass\n\nWith python 3.x, classes can use the metaclass\u003d logic\nto not require usage of the six library.\n\nSubsequent patches will replace other six usages.\n\nChange-Id: Iefdc99c338c7aaea18d535426c4676dbedb44f32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aba14934e7c5197ef7c045f3517c942c0e3bf98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aba14934e7c5197ef7c045f3517c942c0e3bf98c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
