)]}'
{"id":"openstack%2Fmistral~753196","triplet_id":"openstack%2Fmistral~master~I80c713546fcc97391c64e95ef708830632e1ef32","project":"openstack/mistral","branch":"master","topic":"remove-six","hashtags":[],"change_id":"I80c713546fcc97391c64e95ef708830632e1ef32","subject":"Remove six.moves","status":"MERGED","created":"2020-09-22 00:35:18.000000000","updated":"2020-09-22 10:20:15.000000000","submitted":"2020-09-22 10:18:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753196-1600769903438-7e3b2230","meta_rev_id":"d4f11ed670475cb24ef6ddeb3551c3a6c2d2abdf","_number":753196,"virtual_id_number":753196,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-22 02:24:02.000000000","_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 10:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":2,"date":"2020-09-22 05:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-22 05:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2020-09-22 08:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":1,"date":"2020-09-22 08:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 05:13:27.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-09-22 05:35:02.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2020-09-22 08:37:04.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-22 10:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af3be02c60479e4351361026a1fd0145cc53d202","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-22 00:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81567a2936b586857bf9467bb6c13311d0db6b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 02:22:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3062daea578f4bcaaa18fad5063862e4 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc3d6c370e248828a272e28fc14d70e : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00a956b366f54e118abe771980294a96 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5616076c0f24d199c221d184f11dff3 : SUCCESS in 14m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a16e657c6091425780f3a3d5fe6f272a : SUCCESS in 8m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cfb3d28052548f4bed7dfcc2e330c97 : SUCCESS in 9m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0f34c0584a8746c69007fd1ad05ab489 : SUCCESS in 47m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b01c3cc065e044fcb27f10efd7a991d7 : SUCCESS in 47m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f698d1630048b5881aebf2545714de : SUCCESS in 42m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/777fbc5d5a2b49c4bd2b651a8a64b33d : SUCCESS in 41m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/111be9262dc04324bada863e26bea613 : SUCCESS in 31m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a8b90689771e40118d5d4b1da1ce5af6 : FAILURE in 41m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bc8e9ca165b447af9ccba782cd0d2c1d : SUCCESS in 28m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c34f3be0dc9e4f9fbbd4db214346365e : FAILURE in 1h 44m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"973ff7b8f07c3407da35c5dff07d1c1f25b09e4b","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-22 02:24:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8cde52cb27af4f80652905e345def358c1dc37bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 03:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa92b627ab5d41bbb13a7d1a833de46d : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1a2e6823e5b410a8d109dcb9232e680 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc471d80ca1c469db157b03565fc7855 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ed940324324decb1b95d78a47d7322 : SUCCESS in 14m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36fa94cfa86640d092d4d18a4b328dd6 : SUCCESS in 7m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea9f261d62f743feb69490cd63337289 : SUCCESS in 9m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f369f4468e5b43e5ad3dc50efae0f94c : SUCCESS in 43m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b7f3a2293547dcb723ba17b0b9a79f : SUCCESS in 49m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93d2f12bb8a24976b706e1316ceb8f59 : SUCCESS in 41m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8376126fe7e44c4ea6c2922391e0c45c : SUCCESS in 51m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/88d318d0099947c7a017e6fc891bb865 : SUCCESS in 42m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/21db591eef144944b841e8578def177e : SUCCESS in 29m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/64aa270035eb46d0a9d1bb4857433374 : SUCCESS in 26m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/65cd84b417554e4ebacbacb3b1092e8e : SUCCESS in 1h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71e5c043a27736504b220c3e486acae85994e805","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-09-22 05:13:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40a9115d0f46e59aa761cbfdfbbe0e9962550d17","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-09-22 05:35:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f140031226c84c183df6ccbbd15c66055763aa84","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-22 08:37:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7584d5df97783c2ef7bb219d4688bd2ed0e8b0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 08:37:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96db21c751e4bf5c42807c3f63d85b0c6ce15bdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:18:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64bc0dd030464b439fd4d68cc393c504 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7bfb9ab63514fb6bbb38e6d7dcb8e51 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae8ef293ede0431bae38a078b48875be : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c02e93136744fd897dcbe6c5a07ca93 : SUCCESS in 13m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07804dec3d58459db4062befc276bb45 : SUCCESS in 9m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9c7c129fea4448ee801b705aa15d2088 : SUCCESS in 38m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f333f59a7c44357ad1ebbe7ad4a1bbd : SUCCESS in 1h 36m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69c91bc510514737a583953e9e6fc40e : SUCCESS in 42m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e9121708fdff4e8d8ccd2d31226b59ba : SUCCESS in 47m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/589951624343474f95943fd74988cea1 : SUCCESS in 44m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5384d3964c0c4456bbf58586fde703df : SUCCESS in 30m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c9f57a8883a141719ad591ae0c193714 : SUCCESS in 27m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1f262faea87752b49a31b465d8d18ab99a9ae20a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:18:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b27f84db0e2403548d9b53e05741f8048756cb7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:20:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed2b5ab6d5e43899b7f39668812146c : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da5ac25415f20fea6fe92afaca2a4385ca9145dc","revisions":{"da5ac25415f20fea6fe92afaca2a4385ca9145dc":{"kind":"REWORK","_number":1,"created":"2020-09-22 00:35:18.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/96/753196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/753196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/753196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/753196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/753196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/753196/1"}}},"commit":{"parents":[{"commit":"05071c9d9bf61b2ac0a4ba7d8f2c2e057ac0e102","subject":"Merge \"Remove usage of six.add_metaclass\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/05071c9d9bf61b2ac0a4ba7d8f2c2e057ac0e102"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-21 06:49:03.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-22 00:34:20.000000000","tz":480},"subject":"Remove six.moves","message":"Remove six.moves\n\nRemove six.moves Replace the following items with Python 3 style code.\n- six.moves.urllib\n- six.moves.queue\n- six.moves.range\n- six.moves.http_client\n\nSubsequent patches will replace other six usages.\n\nChange-Id: I80c713546fcc97391c64e95ef708830632e1ef32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da5ac25415f20fea6fe92afaca2a4385ca9145dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da5ac25415f20fea6fe92afaca2a4385ca9145dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
