)]}'
{"id":"openstack%2Fmistral~754371","triplet_id":"openstack%2Fmistral~stable%2Fvictoria~I9b0b75ffd5c073a25f8a84bf157cf22988923958","project":"openstack/mistral","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I9b0b75ffd5c073a25f8a84bf157cf22988923958","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-25 12:28:17.000000000","updated":"2020-09-28 08:18:29.000000000","submitted":"2020-09-28 08:13:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754371-1601280806215-e5a30d7c","meta_rev_id":"7abade2e0d5275158e4d59f0a12a76b9d4ebcd91","_number":754371,"virtual_id_number":754371,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 08:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-09-28 07:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-09-28 07:20:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-28 07:20:48.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-09-28 08:13:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48b74e10a63196908f038863fc0b1d1b94c2a455","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-25 12:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e42c38a544167d58bd812dd4ac582adb25c8ee1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 13:08:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/685019230b5e40acba219c6d8b52b6f6 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d7de104c858496487413aea55ee8be8 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/314daf2d24b64fe4b82c9315a11b8361 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1503954ce84a4ed79550333841dc6ff3 : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2e954791d7744d7b084371799e55452 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc881dd164dc4b1da7c2fb02de26cb4f : SUCCESS in 12m 10s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3e78218da306446086e35a40c33b525d : SUCCESS in 30m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c2494b51b8cd44ba9b70b4a96176f4c1 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f11d8607ca77310fc7d7106d044961336623e98f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-09-28 07:20:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4846be6165648c6c929f8d15e84743711060314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 07:21:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fc90d072c0c27dfba136009e3e34a5b5761e8f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 08:13:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85218c8b4264a539713f3951ea32b91 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/985e87bc19c242fc91f76f320d08e1fd : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22e248598ea445d3a970e57abccfb12c : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49807be580374185ba8262a627ceaca4 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d2a85a83f14e6ba83efb64c96fc7a7 : SUCCESS in 8m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d134e943bde84496b04b78ba1502bbee : SUCCESS in 30m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9ebcebad94644f7aab3a02733ce77ea8 : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"77ee9593763d96dc577a6dcede11bac524fe485a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 08:13:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c434fb64a18a240f178b6ab61ce73455e377c8e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 08:18:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff99a0183ccb485faf0a709f406ef85d : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26f52016b6f47a6ff5f15f31ee9bb005727e7267","revisions":{"26f52016b6f47a6ff5f15f31ee9bb005727e7267":{"kind":"REWORK","_number":1,"created":"2020-09-25 12:28:17.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/71/754371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/71/754371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/71/754371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/71/754371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/71/754371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/71/754371/1"}}},"commit":{"parents":[{"commit":"068e5cc0e9ab73a3046a6b8de251079a784b9139","subject":"Add release notes for the action provider transition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/068e5cc0e9ab73a3046a6b8de251079a784b9139"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-25 12:28:12.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-25 12:28:12.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I9b0b75ffd5c073a25f8a84bf157cf22988923958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/26f52016b6f47a6ff5f15f31ee9bb005727e7267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/26f52016b6f47a6ff5f15f31ee9bb005727e7267"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
