)]}'
{"id":"openstack%2Fmistral~755481","triplet_id":"openstack%2Fmistral~master~I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a","subject":"created a new Api endpoints and added dynamic actions","status":"MERGED","created":"2020-10-01 10:36:03.000000000","updated":"2020-11-19 07:02:41.000000000","submitted":"2020-11-19 07:01:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":897,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755481-1605769266609-dfe5b5ea","meta_rev_id":"6469b61c47b97f2e7fe13e544d19f55a7a651388","_number":755481,"virtual_id_number":755481,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 07:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-11-19 05:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-19 05:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-19 05:59:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 05:35:25.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-11-19 05:59:06.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-11-19 07:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18dd31fcbd6eecf34c9477e8ed4f4f7d14c59413","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-01 10:36:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e8298899899f8daee1be5f6af19e627b71fb1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:49:58.000000000","message":"Patch Set 1: Verified-1\n\n(60 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a5c1f468304adcb0b67eda4561e997 : FAILURE in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b79c6af4bb742e08294ede3e4a7cced : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/679dba6cc29340ff9c553120c7d0ada0 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58019439afd14b47b164bcd907338284 : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b26c05e85d3c48eb8ab9b6f541e14165 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07c12bbc1e644d0882cc9c0f1027d596 : SUCCESS in 10m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/23d9da4228ce4efabcf3607d7fcd33a5 : SUCCESS in 50m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75a64da61d354d10aec36e7f60129733 : SUCCESS in 46m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4682f6baa0942ebb4a999f479b5a7bc : SUCCESS in 53m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b5832dd26ca04c7c98b1fce0d7c82143 : SUCCESS in 54m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/951989dcf9494118a636f9b16d5122f5 : SUCCESS in 49m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bca65850b64b4539a88c71e3b2831110 : FAILURE in 29m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5568f278b7a04851ba8cdcd9731ebf9b : FAILURE in 28m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/20d813e427d64bb39fefe7e95a2128ff : TIMED_OUT in 2h 03m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70a11640325ec8720a733fd4ad0feb30e5e4b2e2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-01 13:32:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93773b2cc2f248a301dee5f214ff048678468ea2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-01 14:27:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa2eb06ff42b968c1a81f27339d8d097dbf024bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 18:45:29.000000000","message":"Patch Set 3: Verified-1\n\n(48 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac55b98dfaf841b0a1ab514da2dd9ee5 : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47a4e6accd06405b9a57a5924852d928 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88214fbe4f004990be446c8ed1ca1db9 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735fa94ca46a446e8bd14588cea1dfd1 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb8d274a436c4b609feba0b7b003c151 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ad7e5610054b53affa79e922dc67d8 : SUCCESS in 8m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1923dab066b44b89990306baf09d84fe : SUCCESS in 50m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1946374fa63847a7b6f21ef93dd32cf2 : SUCCESS in 44m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21118a66eb7a43adbb176488306c3985 : SUCCESS in 44m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c922abd7e94945e3b7adbf73905be581 : SUCCESS in 40m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0b6e7bddf77e485a98a59d3b49e4a5ad : SUCCESS in 48m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/89aa9371d1864bc6a93ca562930fad1b : FAILURE in 30m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/058ec97bf4a04ecc9e9cc4630ac26c72 : FAILURE in 28m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/52a1f4cbcf424693a1522287461441e4 : TIMED_OUT in 2h 03m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"106e88ecfbf165da92f05adf4e58129f342afcde","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-06 09:02:51.000000000","message":"Patch Set 3:\n\n(8 comments)\n\nI realize the patch is too early to get reviewed but I left some comments in advance anyway.","accounts_in_message":[],"_revision_number":3},{"id":"928355048e6ae0be12daf5913f6054900ecda03d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-11 06:56:11.000000000","message":"Patch Set 3:\n\nthe names are just placeholders for now,\n\nthis is still very early WIP, these issues will be addressed later down the line.","accounts_in_message":[],"_revision_number":3},{"id":"690c982cfb8ca6bf07a14680e6346ec4aadeeb06","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-13 13:40:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"900e90f33813cb37a49a66e9098d09880427e873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:49:43.000000000","message":"Patch Set 4: Verified-1\n\n(60 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7e34b991f204e92ae36a669fef1bd3a : FAILURE in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e0fdc494314b3e9d9641933857ff49 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9065ef7b27a8411a8b5bb7f4a0c97c16 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7daa7d976a274609922b33111666bb28 : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c77846234fe64629a8b41ffad8496d05 : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/250df073e0de4d0cbff7b223fd717be5 : SUCCESS in 10m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fd9a8e46523c47929ad2fee040f3c702 : SUCCESS in 45m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8293f230105542828a94c18a62e77063 : SUCCESS in 49m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71c8052ad3e149a0be93f90bedde51b9 : SUCCESS in 44m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f66b80161e354e88a18b16dad4b1f0d3 : SUCCESS in 52m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7a01ee7f775542e3be44b4e0590547ef : SUCCESS in 50m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/024dbf20efb949c9861551e8dceb77ef : FAILURE in 32m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f243a9e4857849c3ae6b263d9b9281e0 : FAILURE in 27m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/02077bbc87864112a1b415c0aa1018ac : FAILURE in 2h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a1dc29b80a11f9ee80ffbf639a6ea99639c7b3f8","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-14 15:12:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"76156311ebfee9c1ad1cabd7b6cdad939a0a56d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 17:26:46.000000000","message":"Patch Set 5: Verified-1\n\n(153 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091207b1eee443a095dd793b109fb93e : FAILURE in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ed14142e6fd487b9fc6d83d38e70279 : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/269eab3bf4744cad8183630ef0d69533 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cf4d82262748729fa48ab682039790 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b79fe915579d4d8ebb52f79de8afb2d6 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff5176025a244fcb9af087d1547d7df7 : SUCCESS in 8m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6c72eef1b75f477fae59f6a9f52d555a : SUCCESS in 52m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4afb19e4efed4668b1348d13c8efb3dc : SUCCESS in 55m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea241dcc6d824ab5a2ae02dee56f58e5 : SUCCESS in 48m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bd8c744947284172825d10a5e0333ce8 : SUCCESS in 1h 05m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/964609663cbc4af982e38112435bfac7 : SUCCESS in 50m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c6bbd0fa88fd450691748d35f20a8f62 : FAILURE in 29m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3934cea32ab34123b3a6b088ddb36b73 : FAILURE in 27m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/276f5f5d56944da29deb8365a0eaefd7 : TIMED_OUT in 2h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7ca20dd04ce7f090c75d5b82f2d9962ced8a5aed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-15 05:08:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d248090a5a848214833939743ea4c7e7b5f8336","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-15 14:48:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"27bb03701f6a38739c54121c2e7ed20e9783f0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:01:39.000000000","message":"Patch Set 6: Verified-1\n\n(162 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cf20a883478438c874d4535ff10535e : FAILURE in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dccd43d3231448d68fcbf21fe8d73c18 : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e908196cbfe41af80250a36ee11e3a5 : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82b7c25e93047e4951e65c4c2092533 : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23d6eb06a55a483d879ee85400d337b8 : SUCCESS in 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1812b907bd44e82b3bb8ce999a5966d : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/de0ecabbdd594f549003cfdba0a2f4ce : SUCCESS in 44m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/381f2ceed0e34006b9eaec4d83156694 : SUCCESS in 49m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f937f64ba3a94353aee1d416c5717a69 : SUCCESS in 46m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7077b2086bc742768293d1db94d0a744 : SUCCESS in 46m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a4fe757c6ebc47b395ec87634ff2fda8 : SUCCESS in 44m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b591302a0576403eb4062f2abc87966b : FAILURE in 34m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b3e1ce47906b4a1e8a677cb493f3900f : FAILURE in 33m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7bbabb8ad7fd463fbe7496836097aa6a : TIMED_OUT in 2h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"db7d47280bfa1e7e8466a4cc862933ae8ea62dcc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-16 08:20:36.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ba4477bc2b05bfff61586f252082fa649b7ae446","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-19 14:16:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"822febc8b1480f2a8389a6045ff545248810447e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 16:30:44.000000000","message":"Patch Set 7: Verified-1\n\n(83 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7bde61e74f342208dfb0db9dc1194a2 : FAILURE in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f697f0f7d94ed48ded9f22d56eb6af : FAILURE in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a20a2732782c4e3f9d498ede8ff91144 : FAILURE in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6af9bb49db4c198c9b89948504c453 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a95fcfac2a8149dba3b3a03c39e77bc4 : FAILURE in 11m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1e3a8c454743ada8a67ec048546cdd : FAILURE in 11m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3de1bf3736e543888b7ce96e8a373ccf : SUCCESS in 58m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e44f2b02db134576bbe8634527ab587d : SUCCESS in 55m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ea4595a64824ae08924c2ceea736f70 : SUCCESS in 43m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c84e7eee871149bea13ba822523ca6cd : SUCCESS in 42m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1f45b81d2e2d44eab0622166c7b5c1d5 : SUCCESS in 43m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/21fc210c08414963b9a23d2f4e90ff50 : FAILURE in 31m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d6040954360b4ad497feb9dd24e60183 : FAILURE in 28m 24s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c5dbc0029bd245f2b2282ebc7a467408 : TIMED_OUT in 2h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"524a0580f67fcaf214fd26c599fff95fecf02878","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-21 08:12:54.000000000","message":"Patch Set 7:\n\nLooks better and better. There\u0027s a lot of style issues(look at Zuul\u0027s comments), but I believe you\u0027ll fix them in the final version.","accounts_in_message":[],"_revision_number":7},{"id":"3de6d03ed17a7a3659b56125622f47149c2c04fb","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-22 14:07:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ac3de7fcd5463b6030a5d5267e75ffcac7f1f392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:14:50.000000000","message":"Patch Set 8: Verified-1\n\n(124 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366fe5dcceb342cbb55823d7dfa380e3 : FAILURE in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a861dc57bb384b719b19a048642e6b7d : FAILURE in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b543f3abac04691939374a9e6560d58 : FAILURE in 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4945c836accb400fab483267b09d59f9 : SUCCESS in 15m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53512fe90cd44e01ae13753b0084f446 : FAILURE in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4254e1aca09454fb6a6939c96257a79 : FAILURE in 10m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5d6bd64359042e78c3090c938487268 : SUCCESS in 48m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/144d3e5fb7a94fb5bc1e13dae73e8bb7 : SUCCESS in 48m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c77f972941d9475db005a7e7830cf4a4 : SUCCESS in 49m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b1b8fd18aeca4d1983f728d5e94bff49 : SUCCESS in 59m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/657bee96f24f4c6882e5fb71e86d0b39 : SUCCESS in 51m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/25888380e172422998598f86d4256837 : FAILURE in 30m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7304dcfa7bc04f8e84a1c8f2978978e8 : FAILURE in 29m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ebd9fffa19b9490990a76fa17e8a554e : TIMED_OUT in 2h 04m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8106fe6aff3b683fb6077d898f4723081a16c197","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-23 06:37:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ef5a9f28e7071499eb90a1b3beb726c59d53feba","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-27 15:53:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a3334f78f0f94fb50e98c1e00001780efd54a7d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 18:03:09.000000000","message":"Patch Set 9: Verified-1\n\n(125 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa0d4915d6a4da583b719c433a10306 : FAILURE in 23m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8517729c2f0c4505bc7542c29fe9e72e : FAILURE in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13d87aebe50f439587cba59f8b19ddfb : FAILURE in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9db322e2a2d4e7988666c9796b4ef26 : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76b1efb999974ffdb74e391ac401967f : FAILURE in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e42b32e116c14627b1fd52e3be7c19aa : FAILURE in 8m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a88a5672921b4e9ab8c261cc88733126 : SUCCESS in 52m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1eb948b9920a48989765f8a04f29a350 : SUCCESS in 46m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec301f9ad247462683e025a4d0cf0f4a : SUCCESS in 37m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cecaa5e2b4a442108b355be522ffdc7b : SUCCESS in 52m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d0ddfa38cf7a42e6b24d90ebecdf0124 : SUCCESS in 49m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aceadb557a01479bb9cffcd4edac2335 : FAILURE in 29m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9457424fe597416bab36d3dabbb81720 : FAILURE in 28m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/98ec1f76e11a4a70b255beb451e4db5b : TIMED_OUT in 2h 04m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bc94cbdd4962eeb3851aa7df070b849a4e70eac0","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-28 14:21:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"02db7eab1cf705fe08efd63f545b6ea4ede7235d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-28 15:45:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"15938ab03a5968991350ffdfbdbc283f45274bd9","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-28 15:47:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f82b1ba901ce00955c4c2430fbf270c96b343cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 16:38:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2635828734d9457d9621583dd321ed26 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0282539886724c45833e8e282997f7a5 : FAILURE in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/390c6c7348d0422fa6a9f534f9db733c : RETRY_LIMIT in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a61822e3c2448d08ce76412e590cd7e : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b051a4b0d68c4c02bcce92b271b44f87 : FAILURE in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0475699e6c5459bb74c5e9e129c5a76 : FAILURE in 7m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0bf0c9cb7d66440c9b18f58cdaf920b3 : SUCCESS in 44m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0c27beccdf1487ebdea4b6e694ef15f : FAILURE in 11m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6011ef0f5a094a91860306a2aef58214 : FAILURE in 11m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5b76542f35074c27bde70d9b22dfa837 : FAILURE in 11m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2e361de0f5b74102b0b2fe637bba62ff : FAILURE in 12m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/264d3fe92b2d4295a9b6a4fcefd4e028 : FAILURE in 32m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8a42ad02d1d24fd5b6dd7576b3515e9a : FAILURE in 26m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0a9d17c6ba6745ea9126afaf9a18d21d : FAILURE in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"af6f774dd236ea6ffd7e6b834bc7c9db9a9d4a87","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-29 08:01:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"89449cf5771c4dd13f4be8a8e6c51f0b0fb555cd","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-29 08:23:39.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7b89b6ceeb9279ee61de32b0279c84df4a6c9abe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-29 08:38:24.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"9f0c7ba90bcff37f2edf56294e5f80c3796ea1f2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-29 09:57:23.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9fe04735790546d954de0057bd2d8d24fc07dfbe","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-29 10:15:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ac803a2001a56a8b9e9fecc024430b45be5c255b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:27:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47e3b5db0d046e186ab5a963e600d62 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4870b9734064fa59a928893aaec8359 : FAILURE in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84a7b9db3a1141d9b522236069522d28 : FAILURE in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f768ffd94374e0ebc8f16a649ef15ff : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/208a73584ecf480a9844c67bf8227bc0 : FAILURE in 11m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3605e230aa2444faa33e8333ac542d2 : FAILURE in 11m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b4dc8a7299714402bcb11a6c6baae7b9 : SUCCESS in 49m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bfd2da2b71d4d66906bb4174269e835 : SUCCESS in 49m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e6faa9a05644702b201920f64716b0d : SUCCESS in 55m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/76d23428cc7547789946200438df3478 : SUCCESS in 49m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dd0e0ac8a5bb49ad9a2877c2f98da559 : SUCCESS in 48m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d65cb475ae97402488a19b652c172ea8 : FAILURE in 28m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/52bd0c4b3d5f4478962ed1731ba42129 : FAILURE in 27m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f9753b8619ea41c68add6eb2085c6148 : TIMED_OUT in 2h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"77e5d6609289c11c3a362055b0826da7a8f3c7e4","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-29 13:02:00.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"7bf4c85c9ea2ce33b50d3ce378f3dd0041cfc8c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:14:10.000000000","message":"Patch Set 16: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c60537072554d51bbbe8ec1d9758637 : FAILURE in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4baefc0303c541e5b716ac9d360f8fa6 : FAILURE in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e49f43ef961a4840887eeac307539676 : FAILURE in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448a6c422fdf4816b2f21d4f833967d0 : SUCCESS in 6m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eb0a4fcd2044ec0944aae3826e7ed86 : FAILURE in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8259c70ea8db44c98ba522314c573d4c : FAILURE in 10m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0bf74ca5a17c44069ef241e4c1bd834e : SUCCESS in 46m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62f79d429bc740c78c6fd038f813a883 : SUCCESS in 54m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48b7abbdd8074796b9b25b6f8d581c16 : SUCCESS in 51m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/449bb6fb0f984eeeadc96492987f8e36 : SUCCESS in 54m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7a9544068551432281f5187d2fa45269 : SUCCESS in 1h 01m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4fbaa7d69ad14f258bfe3cc869d9e1ef : FAILURE in 30m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c34b0679303241d297dbfd46abf8ae64 : FAILURE in 27m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/26fd4fa9fee1424b948707ba390c82cc : TIMED_OUT in 2h 04m 07s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c053a082eba934ff5558567793042196ac80ce01","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-10-29 21:09:02.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"00bdc28ad6606b2340bdced75f3827920b0e0d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 23:14:58.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7955a2f16c31499a99cd11b3d001544f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77eed08928dc4004a926801751e9d96e : FAILURE in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad92e968b2154fa29bc9c41930196cf9 : FAILURE in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1c56085c824bf597c87d65eb72d616 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c80866de994fe9809595af30a8371e : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4395ce358cc74e27a0a3c2714bf568d0 : FAILURE in 9m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbaeb05bff4642609e61f8e2668b8280 : FAILURE in 7m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b2c872f0c5b745618781fe5b78060ed6 : SUCCESS in 50m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db688b0e51214c55a7078d219407e786 : SUCCESS in 42m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/581eb4d6b48d42908190083c9af2399f : SUCCESS in 49m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/aa05de033d574f25818ba82d0636e8c9 : SUCCESS in 51m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b680203ff81a412698f76aaea066ca11 : SUCCESS in 48m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3ad8faac28884850bca4372f8465b35a : FAILURE in 31m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d31d5192255f4c0692f20aff244d97ed : FAILURE in 28m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f2d0131eec164eeb90ed5acffe503833 : TIMED_OUT in 2h 03m 43s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d65c85e313b6eb1388cf9a2f21201732db4f2a13","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-02 10:44:12.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"6f9f2bdd0f68a26fba46f4a0ae588e99ddccbff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:59:16.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c0e37b812b4d519dde009b63ab387d : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/412bf0669e924f4592ccae47c3171836 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76e7ecbc3d8d49db826af05a8f021972 : FAILURE in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6a899ba23440bb830308165a84e3fa : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/462fbeec4f2c4a6cade9d544bece4c0c : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99f1287f1d9f4f0185003b8321c0d106 : FAILURE in 8m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb6c723fc1504b48b3a36a24af0c22b6 : FAILURE in 10m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ab704ce056874fbaaa845c71d9b1c87a : SUCCESS in 53m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/868f636537c94e68af3dcf77e082e05e : SUCCESS in 51m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d611ae57e73e451eb879130994c232b2 : SUCCESS in 51m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c383648f5ae94422a3d83688912a3229 : SUCCESS in 49m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a338e430d1334d298b244cac21e2485b : SUCCESS in 54m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6a692049f0274e978483e5077a1bf332 : FAILURE in 30m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4c459bde631842f3b83ac7ba769c237b : FAILURE in 26m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/24a45344d4b2440b8f00edd1223d0060 : TIMED_OUT in 2h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"ad74c3aeb7687f2d44fee0569aaa771996e5f98e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-02 14:51:22.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"c8e4db66cbcd272dcaa242237bf9decf36adc7a4","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-02 15:14:49.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"52c872d751b241c1fe00f2b6964cf8258ae7ce40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 17:34:29.000000000","message":"Patch Set 20: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa65a4825f34428b90582ce1b3203e3e : FAILURE in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61fbab1e462742a892c8bcf02f92fe6e : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8932e95456094d78bdb56240657d453b : SUCCESS in 8m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b7156018fa74864be51a7e93770c078 : FAILURE in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a5cf0a549e4715b03e119b65c8742e : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcd6c96c06134e3193700efc63973baf : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cca237e510f4bac86d021a8992fa196 : FAILURE in 9m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1b9adf547d5401f89dafd3bde46586f : SUCCESS in 8m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/633ab68f7cd940c3856c40957e7cdecb : SUCCESS in 1h 17m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8f6943239f64bba8c52cfa44f0aedc2 : SUCCESS in 52m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c888ffed82447f6a9a4cf6b2e77c7e9 : SUCCESS in 52m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/04a8e14c824841f89cc07d55d973a71b : SUCCESS in 47m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7c7e89e81e5b4b449d1a15854f72a75c : SUCCESS in 47m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/36b9d2563b974110b44bd6dafad8b13b : FAILURE in 28m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6eb012033eb74f5581746c547bb7ebad : FAILURE in 26m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb90474867cb4b3486ef3abbf7bce221 : SUCCESS in 2h 03m 46s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0a3b9a795c7c9f3376fbb6208f494b67fadcf1cc","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-03 10:26:32.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"427ac054d9ba2c82ce7318632f5936c498585bdf","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-03 10:32:22.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"379d0335ce2678b3916dac9cfa808f67efd2e0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:41:31.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd1508918d04dc799db3a2a063d1fc1 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0db5d74c58504759a7551b38ba0ce1d2 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0062cd4624a74ae9b8edafcda917db18 : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b60d1d2e21f473db728919d1b61df2f : FAILURE in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c75306db7464094bfcaa05e3b1c2901 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/109169a420eb4eb9a825f19e3f658d9a : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d53efff36c34f859aead127484cf573 : FAILURE in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38de9cb48ec0440c9239597fadfad5d4 : SUCCESS in 10m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1fdc87cc6a14488687252f422a5183c8 : SUCCESS in 38m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bf53f103071474bb9fb83a00e4ce923 : SUCCESS in 42m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90276a4fe1fc42e69fd97beb4f3425db : SUCCESS in 1h 00m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a9dc2f1825b6421dba2166a82bf164bf : SUCCESS in 1h 04m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5dd8cafc50bf44ed994cfd263e345220 : SUCCESS in 59m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1e48f761fa6040c9a11c86c5a2323d3c : FAILURE in 29m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/49985ba1c8c54bddb7f638fd6f260ffe : FAILURE in 27m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bf832ba62d6342b188fc4fddec95924f : TIMED_OUT in 2h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5df14cf3e1d512e640e06002a6e8cafa6e3ed511","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-03 15:19:56.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"ee450ab1f1d06348b67f176f94691dac4856e113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 17:32:02.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8983f3afca4683b73bbcdc936f900e : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7f71e7ebb49473a995b3832daa2156d : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83063d01a7da4e97b22d1a0fd7a2b330 : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f082e4d05fa4e2ca317ecf172efdb7b : FAILURE in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbcd57c544941ba98c1c714d18a4217 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256ee487239c4352af7863c99fce59f4 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b52dbaa581144eb88a4ee37a9aca1672 : FAILURE in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81d0c646ff00463c8c75f1f5cd6cee50 : SUCCESS in 10m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9ba0af9e37b7441e9ba1dcd22f1629fe : SUCCESS in 57m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d52251d72d8140139c73ab29746dd422 : SUCCESS in 42m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7caa0a10c76e47b3b3f5107177e41777 : SUCCESS in 44m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/269d4e6383024556bfd568132814bc1f : SUCCESS in 47m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c834e64300754c3398051430fa998670 : SUCCESS in 47m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1afd2c0d973d4649ae0785ec6aa594a8 : FAILURE in 34m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c5ac5958944540dab8f1db89d0722860 : FAILURE in 26m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a146e73dcc2467b8cb286231b559a87 : TIMED_OUT in 2h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"8fa9d0e631d3e436ef3701ca32d646024b27c1ca","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-05 08:07:43.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"540e08d160469b866ca2f5fb9dc5fccab937d657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 10:21:41.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada9a4b9a40d48c0a74b465445ab58e5 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8974731adb3148da92402cbc9f5c0db9 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ad40a6e7db2402cb67bc8cd961e65fa : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28382790702144d6bf745a0bc0ef9a10 : FAILURE in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc7f85f553e496f96d304ab7b9a426d : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcbfd9f5ef2345ff86b8527cb68c9be5 : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/275e4246d3234efa9a2519beef4f3394 : FAILURE in 7m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16436bcba7d54d59bb7f99768f506b88 : SUCCESS in 9m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c76a37b5cd41453ba5e6904fbe4fc0a3 : SUCCESS in 1h 03m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d39535ff83704aa0a5e3e996d88920f4 : SUCCESS in 1h 05m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b29a019c3a564baa8b1791a2896f8f5b : SUCCESS in 48m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/129e6f9b66d14d22bc68c175b76919de : SUCCESS in 49m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/667a8ddcc7034f048244d73b3ca6ce57 : SUCCESS in 1h 08m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/27cfe3af0a8d4ce5981898161578a341 : SUCCESS in 29m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c8435818a1043de97071b256e801fac : SUCCESS in 26m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/da0cf64339b94d368c82d3feac837b12 : TIMED_OUT in 2h 05m 14s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"58d69cfe870a89260702cd181e93610097a8896d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-05 10:42:55.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"1fdaf9439c21c153b4cec13e5a84a5396e1a682a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 12:55:06.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5b5c91e9d54e16b9767d27e3ddea5f : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91d02a8d26204f5b930ae0c275cc1fcb : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8fe63e680e49b7b926e3421f184b41 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a495f9ed27847a0a5b58b283f7adf2c : FAILURE in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72dcf27e95514ab0bae43c2eac041662 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba3bb6cdd4264e0bbe333bdf3d2ebaba : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f343bac78b4361a23da305f043d997 : FAILURE in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f021809e8f2c400ba64cad4f8c28d322 : SUCCESS in 8m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b8116403268948898bf190596c3d921e : SUCCESS in 51m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d640a4ba6df4b5a930766b03bcf7a03 : SUCCESS in 47m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c37d228a57541f29f22fa4ee0b7bca9 : SUCCESS in 56m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7e50367f8b4d4b5b84b0122f1216c249 : SUCCESS in 44m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b700619177a6420290df07599198e64a : SUCCESS in 48m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/86855c7c19f34c1ba964251be0f108df : SUCCESS in 31m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/16e98f821df74f68bc596ec31ce3b374 : SUCCESS in 27m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4a744f6525e74ec3b895157c996bf2aa : TIMED_OUT in 2h 03m 25s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"0e3eb60f2999c4131497923fad3975c2678e89e1","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-05 14:43:20.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9b78f48d70aa73c2d67934f2669fedce5dc9c5e2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-05 15:03:58.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"f0e6ba0147a1fd58d24bb22e660b27d927e8f24b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-05 15:23:56.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"f32725d5c41144bb09a830c944731b2b142d10f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 17:35:28.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5f9f815b5e4560889c69726cfd2151 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa0696447b44e49a017aae6de03f31c : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97efd6d84cd54a50bf840d3c56d13a7d : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df69ab59fcc346f5b98a6bbe0320b451 : FAILURE in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd80b14cd794868921948b1f44a4d63 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ee982593ce46899f7ff61e667f305c : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f26ab14a157445c3ba228a931972c441 : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d08dd82d1a1948a7b8b675021b02474b : SUCCESS in 8m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/070ba631f2f848789831197661bd32e7 : SUCCESS in 1h 08m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24fba12f0c8c488490a30818c744db4e : SUCCESS in 52m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2abc20f4de104861a5b77db6c53f521d : SUCCESS in 48m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/215cc356956348aead4a29088e73fa97 : SUCCESS in 53m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/23a0d0dbe6014238b4659de44e6e8846 : SUCCESS in 52m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/53d689f90c7a4a5a849053bf43bb3f20 : SUCCESS in 28m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/82dc6af087ec4f5e9e10ec10bafe15f5 : SUCCESS in 26m 35s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ff2990d06c0a401a8253d18e40ad597a : TIMED_OUT in 2h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"9d85161853aaee387c6b77dae644001f249938f0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-09 06:29:04.000000000","message":"Patch Set 27: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"2016880a6fbfa7b98926921329e1214a8ed1ddeb","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-09 08:58:13.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"2bc93edb987f725e95e78f0cf813cf041e111a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 10:07:06.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99ec2f3df8fa4e2ea98effe838af9905 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927f9e1c18024208a4299338afa61bbc : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24cba8e4b54344c484f2d793001d1b88 : FAILURE in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ae40b6c953a4e37ab39afd9a803b27a : FAILURE in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fd111bd47a94a50a052189bcbdb21fb : FAILURE in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fba3be2badc4c9c8eabd54eb166d753 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c9966d11a224cf7ab1f127d69bb8703 : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee643ef7a1784a47bc3a1ac4c3da71ff : FAILURE in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1820976687644cac9c753c0e4a7a7448 : FAILURE in 7m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7ef0d1ff0cfe48779b7b139421fe983a : FAILURE in 51m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/822582b6aca247b6924ee4932bd8f04a : FAILURE in 1h 01m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b0a687632d44fa977bd1d2fda4acf7 : FAILURE in 51m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b802691aedb1426f87c80cf52c0a2d90 : FAILURE in 52m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fe719d81bd9648aeabede1dd10ee0b25 : FAILURE in 50m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1fbfaa2d712f48dca6dd251d03212cb0 : FAILURE in 21m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38892e940ef04970bbec8e2b75c1a5bb : FAILURE in 20m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7cc429433a6445af9bae1050d84dd051 : FAILURE in 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"d48495c3195a40ee6102915b82b94d973b20322c","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-09 16:05:47.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"dc2211ef49489cc17342542420bbe37727fe1e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 18:17:48.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0bc0319a323643d4a850a4abe035039b : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555dae89452348c585610d3dc265f221 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7a757e3986247da9bf98f0c85d22ad9 : FAILURE in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dfe75b536da4685a744f412da5e66dd : FAILURE in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bb6785548be454aa84a995ff2fa0c89 : FAILURE in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb611b97bb74c8c914476ad750b2bf7 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece79b504cf348e38cac73f30b1314c7 : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32efa93b425840de90814158efd526bc : FAILURE in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cff1bc93c6354d5b902724688d46f6c2 : FAILURE in 9m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2508c5ed6b8d4655a06d06307ddd3e89 : FAILURE in 1h 15m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7986c5c9be1040e6b3b9f3889ea96f69 : FAILURE in 43m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b05bf3d998fd47ab8543bdaf5ff791ee : FAILURE in 1h 03m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e8964b7b268f48cbaf6decebcf3b5eb5 : FAILURE in 1h 03m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dfbc1d9900684da4b9ac79095c5277ad : FAILURE in 1h 03m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3f6732c546804b9a97703197dc480cf2 : FAILURE in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/081b1ee63aeb45dc949ce232f75257fe : FAILURE in 26m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3ded94782d064f9f85e2502568caab2c : TIMED_OUT in 2h 04m 19s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"fc6b4f31ecb1564fcb380234da3b3249f2af1f0a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-10 10:36:32.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"e33941e0d1e24aa93f7bd2d84a6f2ad99aa824fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 12:54:30.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3297b7ca7ba2489fb7c790767fd79065 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b35f524d35141f48c4c64f0bd115995 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d1347e1dfb64d6eaae781ed354c2fe3 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4db40a822a3b47f488a592cf699b6c4e : SUCCESS in 13m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/404034e555d44ade9888b6707f1880e2 : FAILURE in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c54498c0ac4f4096b038ff9ce0ddaa : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/829e8b1e5e6847c581be3d411f7dc91d : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0951a59ec9e4334b35a383069eb527a : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e14cd5d4e0704912877cc91fd70c8c78 : SUCCESS in 8m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bffaadef5c8e4c40be096ecbc76f3e15 : SUCCESS in 42m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fc801b22b704c5b9c53f8b3c7526f68 : SUCCESS in 46m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06a7cfd4194140e5aec6668b73d6d8eb : SUCCESS in 38m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/825ec4b4049942a8bda2880ce7a1e75b : SUCCESS in 38m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a7efb217816c4f22b174bdfaac4ea261 : SUCCESS in 53m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b1f76f8ec4c54f4e987c64c748606596 : SUCCESS in 29m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dcd43566363b4d0194ec7e25a1ae59b8 : SUCCESS in 26m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ab0b406806fd4f588b6b3068119dc965 : SUCCESS in 2h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"3cd37779238c53c1444a3eb9c3a604e5ce48cc50","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-10 14:00:31.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"61f1a84fc887110b86a723da46247b685c4c8ced","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-11-17 07:54:52.000000000","message":"Patch Set 30: Code-Review-1\n\n(51 comments)\n\nAli, there\u0027s still a lot of style and logical issues in the patch. Please polish it further. Carefully address all comments one by one.","accounts_in_message":[],"_revision_number":30},{"id":"0c0a9aa877ea37a49a8800a542dc035059da95f9","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-18 10:13:13.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"14e177a09a4161ff26372707d0c01882c6759d83","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-18 10:15:37.000000000","message":"Patch Set 32: Commit message was updated.","accounts_in_message":[],"_revision_number":32},{"id":"4b0e84e2f991a2feb28b52d45de3a4c363499dc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 12:43:38.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3c0775ee439c400c9036b23dcf833a16 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eced805e66d4b63a980d522168c38f9 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00a9e707819847d09d061bf3823334dd : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5ed642b56c2423e8a72c5ddbe0ad8e2 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae46380820c148b38ccb1672c6808439 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e484095d4ff64e6a946d2e05fa74666f : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d5e96b654464a689e1d522dd0c50e05 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e2d38f0c1344985a0359cf1f07bb150 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/612352fb949946a8afa0104fdee5990d : SUCCESS in 8m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5d6fc123f6254e14877734ce9eaae487 : SUCCESS in 47m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff3ec0942d2432e8bba4212835ca13f : SUCCESS in 51m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b28768e9519a42968ec37ccdbc3bda30 : SUCCESS in 52m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5973467a53d643e7b2cc3c3b119216f1 : SUCCESS in 47m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6d4e2097c7c543c5a8d9277e4ba4e4d3 : SUCCESS in 43m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c0a4347e1f8f4024997d26c3670b4695 : SUCCESS in 28m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cae44fdb3ec744ad84b30cf710bc42d0 : SUCCESS in 26m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/886d113bc85846fdbe2b0a79d1637215 : TIMED_OUT in 2h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"02f7ab250baf006237003fd218249989c56398ed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-11-19 05:35:25.000000000","message":"Patch Set 32: Code-Review+2","accounts_in_message":[],"_revision_number":32},{"id":"ed48879c646a4f8b431a9946ed029e535e0550f1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-19 05:59:06.000000000","message":"Patch Set 32: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":32},{"id":"9d55eebc24b3fcbe1bb60e7f84dabed3453ae56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 05:59:21.000000000","message":"Patch Set 32: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":32},{"id":"e531c3dff93147f14131133a8f26c83d5f847f08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:01:06.000000000","message":"Patch Set 32: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52bc476a39d645578669db122f576678 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9230cf2a0973487f81f725dd495bfd32 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32c98ca4f36746d8bfb0b5ff1beb1a64 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ff784362de74773b2edaec34237fbdd : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e859a24d8594a6ba6f7d8a561512699 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd69738425624d5696f2ee3c1ae4337b : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5306482687e54c1ba392179ae5cd8ee8 : SUCCESS in 8m 54s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/df965c593455420cb89bce1c01b8ea0e : SUCCESS in 54m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46b58951378943aeb0f2bf5050425d91 : SUCCESS in 51m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d6f8fd9370e4988afc4e4f3614ef825 : SUCCESS in 53m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c26a285c78824d828d31696e3cb073b5 : SUCCESS in 52m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c6f94d313d33467c880c8e31c3e9ce01 : SUCCESS in 46m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4c487979487c4058b73895cbcf241f4e : SUCCESS in 28m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dd7394a75b134ddfb6c84313121783bf : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":32},{"id":"6865f8fc70cd583bf5fc68495ad49b627f6decb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:01:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":32},{"id":"6469b61c47b97f2e7fe13e544d19f55a7a651388","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:02:41.000000000","message":"Patch Set 32:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004b125efe514b1ca1b400ae638cec84 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52479e687fa84f699c73543441b1f04f : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":32}],"current_revision_number":32,"current_revision":"9be4f8e11937eef91365fb15b224e28ba14516e5","revisions":{"d3135e9df453902d2dc50e249a9d2119587affc9":{"kind":"REWORK","_number":1,"created":"2020-10-01 10:36:03.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/1"}}},"commit":{"parents":[{"commit":"175e5fd5877ba062511fd41fb73261faca35b15e","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/175e5fd5877ba062511fd41fb73261faca35b15e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d3135e9df453902d2dc50e249a9d2119587affc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d3135e9df453902d2dc50e249a9d2119587affc9"}]},"branch":"refs/heads/master"},"14140de52e57f5a55c57736b6a15a233a88dde8d":{"kind":"REWORK","_number":2,"created":"2020-10-01 13:32:35.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/2"}}},"commit":{"parents":[{"commit":"175e5fd5877ba062511fd41fb73261faca35b15e","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/175e5fd5877ba062511fd41fb73261faca35b15e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 13:32:24.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14140de52e57f5a55c57736b6a15a233a88dde8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14140de52e57f5a55c57736b6a15a233a88dde8d"}]},"branch":"refs/heads/master"},"2f6fb1727b368bce7a41dc55b6f956374c16e995":{"kind":"REWORK","_number":3,"created":"2020-10-01 14:27:23.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/3"}}},"commit":{"parents":[{"commit":"175e5fd5877ba062511fd41fb73261faca35b15e","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/175e5fd5877ba062511fd41fb73261faca35b15e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 14:27:10.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2f6fb1727b368bce7a41dc55b6f956374c16e995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2f6fb1727b368bce7a41dc55b6f956374c16e995"}]},"branch":"refs/heads/master"},"993d4769fe000b7694fb13018695adf1f36d855c":{"kind":"REWORK","_number":4,"created":"2020-10-13 13:40:41.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/4"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-13 13:40:28.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/993d4769fe000b7694fb13018695adf1f36d855c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/993d4769fe000b7694fb13018695adf1f36d855c"}]},"branch":"refs/heads/master"},"6ce36747465c5842667a19be7c1c8d696f9ac2f9":{"kind":"REWORK","_number":5,"created":"2020-10-14 15:12:45.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/5"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-14 15:12:22.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ce36747465c5842667a19be7c1c8d696f9ac2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ce36747465c5842667a19be7c1c8d696f9ac2f9"}]},"branch":"refs/heads/master"},"a0397211abc31ad08d399b89926ab63652cdde7a":{"kind":"REWORK","_number":6,"created":"2020-10-15 14:48:46.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/6"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-15 14:48:36.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a0397211abc31ad08d399b89926ab63652cdde7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a0397211abc31ad08d399b89926ab63652cdde7a"}]},"branch":"refs/heads/master"},"631fd03182cc4b3ac707e8abed030498d053b050":{"kind":"REWORK","_number":7,"created":"2020-10-19 14:16:28.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/7"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-19 14:16:10.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/631fd03182cc4b3ac707e8abed030498d053b050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/631fd03182cc4b3ac707e8abed030498d053b050"}]},"branch":"refs/heads/master"},"9f4417658401cdef1507aa981eae8648ddba7fd9":{"kind":"REWORK","_number":8,"created":"2020-10-22 14:07:21.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/8"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-22 14:07:10.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f4417658401cdef1507aa981eae8648ddba7fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f4417658401cdef1507aa981eae8648ddba7fd9"}]},"branch":"refs/heads/master"},"2865864fe8f5e200243f3ee9c55278dbe17ac06a":{"kind":"REWORK","_number":9,"created":"2020-10-27 15:53:09.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/9"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-27 15:52:58.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2865864fe8f5e200243f3ee9c55278dbe17ac06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2865864fe8f5e200243f3ee9c55278dbe17ac06a"}]},"branch":"refs/heads/master"},"22a93b6d27294fac6f6909a54121aee1cdae2e7d":{"kind":"REWORK","_number":10,"created":"2020-10-28 14:21:26.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/10"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-28 14:21:13.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22a93b6d27294fac6f6909a54121aee1cdae2e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22a93b6d27294fac6f6909a54121aee1cdae2e7d"}]},"branch":"refs/heads/master"},"916932bf56888ebc38f2bb1bf414405415e05321":{"kind":"REWORK","_number":11,"created":"2020-10-28 15:45:55.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/11"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-28 15:45:43.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/916932bf56888ebc38f2bb1bf414405415e05321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/916932bf56888ebc38f2bb1bf414405415e05321"}]},"branch":"refs/heads/master"},"37c4627d4a7f1285af75aff52d9d043d0cc51195":{"kind":"REWORK","_number":12,"created":"2020-10-28 15:47:34.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/12"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-28 15:47:21.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/37c4627d4a7f1285af75aff52d9d043d0cc51195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/37c4627d4a7f1285af75aff52d9d043d0cc51195"}]},"branch":"refs/heads/master"},"54d937761fe2394b6be74ca09fbe2fc1dd01b16c":{"kind":"REWORK","_number":13,"created":"2020-10-29 08:01:17.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/13"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-29 08:01:03.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54d937761fe2394b6be74ca09fbe2fc1dd01b16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54d937761fe2394b6be74ca09fbe2fc1dd01b16c"}]},"branch":"refs/heads/master"},"1ee7c3dfccefaf2314ef28ca3ec51eeb8d3b294a":{"kind":"REWORK","_number":14,"created":"2020-10-29 09:57:23.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/14"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-29 09:57:08.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ee7c3dfccefaf2314ef28ca3ec51eeb8d3b294a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ee7c3dfccefaf2314ef28ca3ec51eeb8d3b294a"}]},"branch":"refs/heads/master"},"2d9ef474ef9dd1b894b04c60b1fd444920bbc34e":{"kind":"REWORK","_number":15,"created":"2020-10-29 10:15:45.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/15"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-29 10:15:07.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d9ef474ef9dd1b894b04c60b1fd444920bbc34e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d9ef474ef9dd1b894b04c60b1fd444920bbc34e"}]},"branch":"refs/heads/master"},"0f5cc3626aa6530b96440a1f17c96a5aa47c1b69":{"kind":"REWORK","_number":16,"created":"2020-10-29 13:02:00.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/16"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-29 13:01:44.000000000","tz":0},"subject":"(WIP) created a new Api endpoint for action creation","message":"(WIP) created a new Api endpoint for action creation\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f5cc3626aa6530b96440a1f17c96a5aa47c1b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f5cc3626aa6530b96440a1f17c96a5aa47c1b69"}]},"branch":"refs/heads/master"},"cdfafa155543f76da4e5711e43850eaff84178a1":{"kind":"REWORK","_number":17,"created":"2020-10-29 21:09:02.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/17"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-29 21:08:31.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cdfafa155543f76da4e5711e43850eaff84178a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cdfafa155543f76da4e5711e43850eaff84178a1"}]},"branch":"refs/heads/master"},"1ac0a053b62d2f6799ec865fdbdf09f22da276e9":{"kind":"REWORK","_number":18,"created":"2020-11-02 10:44:12.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/18"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-02 10:44:02.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ac0a053b62d2f6799ec865fdbdf09f22da276e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ac0a053b62d2f6799ec865fdbdf09f22da276e9"}]},"branch":"refs/heads/master"},"29c32e55222d59f1b2517c04bcef41a527bd5f75":{"kind":"REWORK","_number":19,"created":"2020-11-02 14:51:22.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/19"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-02 14:51:02.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/29c32e55222d59f1b2517c04bcef41a527bd5f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/29c32e55222d59f1b2517c04bcef41a527bd5f75"}]},"branch":"refs/heads/master"},"0281cc0ef298adb45cf2f3b0c31afc56698a274b":{"kind":"REWORK","_number":20,"created":"2020-11-02 15:14:49.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/20"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-02 15:14:37.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0281cc0ef298adb45cf2f3b0c31afc56698a274b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0281cc0ef298adb45cf2f3b0c31afc56698a274b"}]},"branch":"refs/heads/master"},"5112e913788060f45b3df2eaf8661fb285435e9c":{"kind":"REWORK","_number":21,"created":"2020-11-03 10:26:32.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/21"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-03 10:26:20.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5112e913788060f45b3df2eaf8661fb285435e9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5112e913788060f45b3df2eaf8661fb285435e9c"}]},"branch":"refs/heads/master"},"1accbe554034e92b14aa6227d5de2e1305c9a286":{"kind":"REWORK","_number":22,"created":"2020-11-03 10:32:22.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/22"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-03 10:32:06.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1accbe554034e92b14aa6227d5de2e1305c9a286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1accbe554034e92b14aa6227d5de2e1305c9a286"}]},"branch":"refs/heads/master"},"cbd7f755e55cffd2dad9c636bef91777a63d467a":{"kind":"REWORK","_number":23,"created":"2020-11-05 08:07:43.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/23"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-05 08:07:28.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbd7f755e55cffd2dad9c636bef91777a63d467a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbd7f755e55cffd2dad9c636bef91777a63d467a"}]},"branch":"refs/heads/master"},"1c669ce1f639da7905d0db8ec5ac1adf577a2147":{"kind":"REWORK","_number":24,"created":"2020-11-05 10:42:55.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/24"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-05 10:42:40.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c669ce1f639da7905d0db8ec5ac1adf577a2147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c669ce1f639da7905d0db8ec5ac1adf577a2147"}]},"branch":"refs/heads/master"},"36372d5164a41410405d26a4d5bffbb366020254":{"kind":"REWORK","_number":25,"created":"2020-11-05 14:43:20.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/25"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-05 14:42:59.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/36372d5164a41410405d26a4d5bffbb366020254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/36372d5164a41410405d26a4d5bffbb366020254"}]},"branch":"refs/heads/master"},"1ffa1aadc3ef1ad51a2beb56d7df646087f0b4d3":{"kind":"REWORK","_number":26,"created":"2020-11-05 15:03:58.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/26"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-05 15:03:47.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ffa1aadc3ef1ad51a2beb56d7df646087f0b4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1ffa1aadc3ef1ad51a2beb56d7df646087f0b4d3"}]},"branch":"refs/heads/master"},"c8f7b350444881c5156bd315119388b0b424a362":{"kind":"REWORK","_number":27,"created":"2020-11-05 15:23:56.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/27"}}},"commit":{"parents":[{"commit":"a8049a24695578c246713fb22977558203306e2f","subject":"Merge \"Address a TODO comment in the legacy action provider test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a8049a24695578c246713fb22977558203306e2f"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-05 15:23:43.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c8f7b350444881c5156bd315119388b0b424a362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c8f7b350444881c5156bd315119388b0b424a362"}]},"branch":"refs/heads/master"},"4cc1d9070d0684d44c98768213d67c90d44e12a1":{"kind":"REWORK","_number":28,"created":"2020-11-09 08:58:13.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/28"}}},"commit":{"parents":[{"commit":"2cd445587368755e7846061728b10ead86790c60","subject":"[doc]Fix invalid url link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cd445587368755e7846061728b10ead86790c60"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-09 08:57:59.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4cc1d9070d0684d44c98768213d67c90d44e12a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4cc1d9070d0684d44c98768213d67c90d44e12a1"}]},"branch":"refs/heads/master"},"148925e7d475d2821928ad396b50c1bc838e7481":{"kind":"REWORK","_number":29,"created":"2020-11-09 16:05:47.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/29"}}},"commit":{"parents":[{"commit":"2cd445587368755e7846061728b10ead86790c60","subject":"[doc]Fix invalid url link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cd445587368755e7846061728b10ead86790c60"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-09 16:04:06.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/148925e7d475d2821928ad396b50c1bc838e7481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/148925e7d475d2821928ad396b50c1bc838e7481"}]},"branch":"refs/heads/master"},"d06dd328fd8aaa9a9237be90b2f05bd890f97de5":{"kind":"REWORK","_number":30,"created":"2020-11-10 10:36:32.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/30"}}},"commit":{"parents":[{"commit":"2cd445587368755e7846061728b10ead86790c60","subject":"[doc]Fix invalid url link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cd445587368755e7846061728b10ead86790c60"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-10 10:36:02.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d06dd328fd8aaa9a9237be90b2f05bd890f97de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d06dd328fd8aaa9a9237be90b2f05bd890f97de5"}]},"branch":"refs/heads/master"},"64b89d393ec48216193996c2830741d12b162edf":{"kind":"REWORK","_number":31,"created":"2020-11-18 10:13:13.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/31"}}},"commit":{"parents":[{"commit":"1f464e616efa2d528549cec484270d6b1e1f7f78","subject":"Merge \"Update TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f464e616efa2d528549cec484270d6b1e1f7f78"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-11-18 10:12:49.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamically added actions:\n     these actions are created and used in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64b89d393ec48216193996c2830741d12b162edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64b89d393ec48216193996c2830741d12b162edf"}]},"branch":"refs/heads/master"},"9be4f8e11937eef91365fb15b224e28ba14516e5":{"kind":"NO_CODE_CHANGE","_number":32,"created":"2020-11-18 10:15:37.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/81/755481/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/755481/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/755481/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/755481/32"}}},"commit":{"parents":[{"commit":"1f464e616efa2d528549cec484270d6b1e1f7f78","subject":"Merge \"Update TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f464e616efa2d528549cec484270d6b1e1f7f78"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-10-01 10:35:39.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-11-18 10:15:26.000000000","tz":0},"subject":"created a new Api endpoints and added dynamic actions","message":"created a new Api endpoints and added dynamic actions\n\n * added dynamic actions:\n     these actions are created and modified in runtime,\n     each action needs a code source to be imported from and a\n     class name.\n\n - there are 2 new endpoints:\n    - /v2/code_sources/:\n       used to add new code sources to mistral.\n    - /v2/dynamic_actions/:\n      used to add dynamic actions to mistral in runtime\n\n - a new Action provider (DynamicActionProvider) was added:\n    it provides the actions created from the dynamic actions api.\n\nChange-Id: I9fe8c28ffdef71016d9dc13aea60a288c8ebaa0a\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9be4f8e11937eef91365fb15b224e28ba14516e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9be4f8e11937eef91365fb15b224e28ba14516e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
