)]}'
{"id":"openstack%2Fmistral~756017","triplet_id":"openstack%2Fmistral~master~I3ec465d5d9505a93850494d767668c021ef63847","project":"openstack/mistral","branch":"master","topic":"fix_legacy_action_provider_test","hashtags":[],"change_id":"I3ec465d5d9505a93850494d767668c021ef63847","subject":"Address a TODO comment in the legacy action provider test","status":"MERGED","created":"2020-10-05 09:07:35.000000000","updated":"2020-10-06 07:46:35.000000000","submitted":"2020-10-06 07:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756017-1601970270099-d979b326","meta_rev_id":"59a6b6d67acdb54dea98c02531360a0b87d28b4c","_number":756017,"virtual_id_number":756017,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 07:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-10-06 06:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-10-05 12:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-10-06 06:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-10-05 11:43:24.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-10-05 11:43:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-05 11:43:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 11:43:24.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-10-05 11:43:27.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-10-05 11:43:31.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-10-05 12:14:14.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-10-06 06:42:48.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-10-06 07:44:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"133a45614a916bb3f984d34c09e9c991af78bdeb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-05 09:07:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00914b11434fd7eba628a5eaaa494db3f8a57c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 11:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49a122ed35245f1b788e7f1cb3aedba : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81f6495833a64ebc8206b063399da4e1 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b5847eee5674d8397b18d72e003c4b4 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c2d310243e6428b8c55e57961b53748 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4306eed2fb44d33a3326a8ade2a45f6 : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f8a6094996c49d38d19e618d686eaf2 : SUCCESS in 9m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9a623543b1e144cbac0e155fcca6c31b : SUCCESS in 49m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79d847d356da4cf3a3ee43919aad071b : SUCCESS in 40m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe3afa2c9c704936ad0cbbda79007593 : SUCCESS in 1h 03m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/160f5fc1cb4e400f880fc5585e5055e1 : SUCCESS in 49m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0a3b500917ed4d939acdb41ef9bc5b62 : SUCCESS in 45m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/eef652ebd540472cafdc6f76bdbad749 : SUCCESS in 30m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/212d30c00b344ecfaf580e8b77b5ca69 : SUCCESS in 28m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7d26092fb1cb46feb4e93808df83d99d : TIMED_OUT in 2h 04m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2982555739829b3528c5d2626d1d0c73e895dd16","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-10-05 12:14:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc5be030b1a03b9aaf0e13e3b4315f93df9c8546","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-10-06 06:42:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e0fbaa94917a693e1f823fed1cc193a061d8750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:43:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3af6695a4d1a98addcd0740668b86aad660d947e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:44:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17348154627d4412922f09a0b4af2107 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dc28a7a1d4345e1b63be48e1ab5e14b : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/556a21ca198c4a81b0ac3c84fde67da2 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ec95f9019443bea806d629bc0b17de : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ce213fc72924479bfbff4f416388ec8 : SUCCESS in 8m 17s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9c0c40b3ab674979bd5b37b0ed283c06 : SUCCESS in 44m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c01f501f42da4e6895bf8cdcf67c0a4c : SUCCESS in 52m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa0158005b24a90b793d89d4f7f60a3 : SUCCESS in 46m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1503700889f34fdb928d4003709c5133 : SUCCESS in 45m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4fc8b4080df84f26b9dfcad65774bec2 : SUCCESS in 45m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c50a9272400e497dbb53a649061e3bf3 : SUCCESS in 30m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6b983fa616354220b6ccdcffe83136f6 : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"01db9686716cb402b931d41f1398081a1bfdf95d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:44:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b8cdfdd939b7271093cec8640c1810989254fd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827b6692d6294dddb19da6edc5727864 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d70823c6a84371c78294c5f43f11944e01bff38","revisions":{"1d70823c6a84371c78294c5f43f11944e01bff38":{"kind":"REWORK","_number":1,"created":"2020-10-05 09:07:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/17/756017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/17/756017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/17/756017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/17/756017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/17/756017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/17/756017/1"}}},"commit":{"parents":[{"commit":"175e5fd5877ba062511fd41fb73261faca35b15e","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/175e5fd5877ba062511fd41fb73261faca35b15e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-10-05 09:06:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-10-05 09:06:07.000000000","tz":420},"subject":"Address a TODO comment in the legacy action provider test","message":"Address a TODO comment in the legacy action provider test\n\n* Added a test action generator to the test that checks the config\n  option \"load_action_generators\".\n\nChange-Id: I3ec465d5d9505a93850494d767668c021ef63847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1d70823c6a84371c78294c5f43f11944e01bff38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1d70823c6a84371c78294c5f43f11944e01bff38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
