)]}'
{"id":"openstack%2Fmistral~756020","triplet_id":"openstack%2Fmistral~master~Ic4b6a9a7ca2784a892d2998359edb220ff8c8911","project":"openstack/mistral","branch":"master","topic":"remove_action_factory_module","hashtags":[],"change_id":"Ic4b6a9a7ca2784a892d2998359edb220ff8c8911","subject":"Remove the module mistral/actions/action_factory.py","status":"MERGED","created":"2020-10-05 09:19:01.000000000","updated":"2020-10-06 07:43:23.000000000","submitted":"2020-10-06 07:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756020-1601969992227-7963eb8c","meta_rev_id":"1abde8ab040112350384b007c946badbce521a50","_number":756020,"virtual_id_number":756020,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 07:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-10-06 06:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-10-05 12:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-10-06 06:41:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-10-05 11:43:02.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-10-05 11:43:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-05 11:42:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 11:42:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-10-05 11:43:02.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-10-05 11:43:08.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-10-05 12:14:56.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-10-06 06:41:57.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-10-06 07:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd1038c7073f04201962a427863f4a69eba7f46c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-05 09:19:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7155de5f1d8a0a00df601abdcb410b86796caa47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 11:33:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb802b045a04199aa12ba7c7670b96f : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b196e6ea133040f5b12ca5fe2aa42d24 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df0c1d12859460b85957a7be9ac1d2a : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d226ef7e7fe14e538f7c8452bae2f34a : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61c18e22e9a143e6a19848308fa6fbb6 : SUCCESS in 8m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fdccb6a9a6c46e9be042387aac7256c : SUCCESS in 9m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/015fc6d4e82249b183cbd2bb2e1e4c9d : SUCCESS in 46m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4d61ec1455447f9afa8c1c6bfb431a8 : SUCCESS in 54m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97d2bef357684eafa1445fb53a41f6f7 : SUCCESS in 52m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5642fa6d3f8343b3bebee3bff41ac445 : SUCCESS in 52m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a85436ce354f4648b2e415f6b7140c56 : SUCCESS in 46m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/119a26c768aa4f83bacfecfed4ada5e1 : SUCCESS in 29m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ee2ae33f16a848ac87d2c8c3f341caf2 : SUCCESS in 28m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6a462073dae6498fb59ac078ac914f35 : TIMED_OUT in 2h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70070402ef461cc400839d369a4f40b314ab24ec","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-10-05 12:14:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c342928d396f43b3b43117980ae3338fe3c25852","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-10-06 06:41:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e49f032989eb1d1c93e087c730c6b69d706592e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:42:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9498cc717552424c5e9f7c8cb26e22d46d5341f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:39:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c42e046f034f10a10cb992d8ca28a2 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/071731a37c1a4c7c90f4f316b6226a3e : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93fc5cc436b8485197e6072aa37e0fba : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1621520880418aa70bfb325b4972a8 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e987cff9c9b4df99c2dc06f3fe16edc : SUCCESS in 8m 30s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c07f9202adad4053ae7127467d0e1b32 : SUCCESS in 33m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1945d4ba40741bca232fa8fe94d35b1 : SUCCESS in 42m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc9dec9a745d49f0a58d9a95030175a8 : SUCCESS in 46m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5ab51bb8712442ec8f933ee073f531c1 : SUCCESS in 43m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e869d10155b84616ad23d64bd88d88d7 : SUCCESS in 50m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cb2b8cdba7e541e3a3356db182bcb939 : SUCCESS in 29m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c6f2effada054a29839325893b663f19 : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b355c513457239ba5bee3889579df710dbf90517","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:39:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e84647ec9198c43f3488e4ebd0b46f23cad90ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33795aaff0db418186a76b6be0376db7 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d81dc75a3c56928c430217ee315aa14fc0ddbca1","revisions":{"d81dc75a3c56928c430217ee315aa14fc0ddbca1":{"kind":"REWORK","_number":1,"created":"2020-10-05 09:19:01.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/20/756020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/20/756020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/20/756020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/20/756020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/20/756020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/20/756020/1"}}},"commit":{"parents":[{"commit":"175e5fd5877ba062511fd41fb73261faca35b15e","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/175e5fd5877ba062511fd41fb73261faca35b15e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-10-05 09:15:13.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-10-05 09:15:13.000000000","tz":420},"subject":"Remove the module mistral/actions/action_factory.py","message":"Remove the module mistral/actions/action_factory.py\n\n* This module was always a weird entity in the system having just\n  one function that essentially creates a dynamic class. It was\n  created just because we didn\u0027t understand where else to put this\n  function. But now after the action provider refactoring we don\u0027t\n  need it anymore. Action instantiation is now a responsibility of\n  action descriptor classes.\n\nChange-Id: Ic4b6a9a7ca2784a892d2998359edb220ff8c8911\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d81dc75a3c56928c430217ee315aa14fc0ddbca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d81dc75a3c56928c430217ee315aa14fc0ddbca1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
