)]}'
{"id":"openstack%2Fmistral~756032","triplet_id":"openstack%2Fmistral~master~I45b97b08184c8d5a88bcc537fb5b1e538f105554","project":"openstack/mistral","branch":"master","topic":"remove_action_spec_todo_comment","hashtags":[],"change_id":"I45b97b08184c8d5a88bcc537fb5b1e538f105554","subject":"Remove a TODO comment about saving an action spec","status":"MERGED","created":"2020-10-05 10:26:57.000000000","updated":"2020-10-06 07:41:43.000000000","submitted":"2020-10-06 07:38:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756032-1601969920840-d979b326","meta_rev_id":"ab4cf3adacb13d2072031785525185712d4914e1","_number":756032,"virtual_id_number":756032,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 07:38:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-10-06 06:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-10-05 12:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-10-06 06:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-10-05 11:43:54.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-10-05 11:44:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-05 11:44:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 11:43:54.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-10-05 11:44:05.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-10-05 11:44:20.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-10-05 12:15:27.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-10-06 06:40:56.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-10-06 07:38:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e39a65346964d8ada7d406199229987c9adc765","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-10-05 10:26:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8450d5775346151876870dc740f8b40dd98a74d","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-10-05 12:15:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9a13e0cbca148fbafb1b659a1bd98012ba4eb42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:46:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4c6d37d8064e2dac2c9d81606b8187 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d6a90a67df42828e60c641d3e6301d : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c9d305723e642e88179f7cc59fda450 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb66acdb0f4d41f6b64c5fbbba814acd : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fa49afac3744d77be8e5c32a73b1c65 : SUCCESS in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/018cc75db9c54631834be3b97847b0a6 : SUCCESS in 9m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/aece2c26288d4a099394b1297c0b8466 : SUCCESS in 48m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e6eef8c08034099a9688c3ad936763f : SUCCESS in 47m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79cc4cad484a4fc0944422e700c683de : SUCCESS in 38m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/50aa64c11a7f478e99a44cbd7a532b8f : SUCCESS in 50m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c7983d9d6a1647e8be9a815671a8557d : SUCCESS in 47m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a2f80eb16d5644fca029ab96e01832aa : SUCCESS in 29m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0cc8f739278c40a490b950cda16e6684 : SUCCESS in 26m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/412642bf6e8d49b5b0377cc673d0ad7b : TIMED_OUT in 2h 06m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad807c6881538206cfb206e69dd6bd6b4792dfec","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-10-06 06:40:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85cdaf2a0a810040c60ee2c87c2aa459c646c5a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:41:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2de400835e46beaeb164f6b9bba6bc3f669a0f4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:38:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1651e51dfc8849978fd89cf97b79b661 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a33fcd97deb74aa8a01d162aa09ed4f8 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35128fc9dbe147ff88b3d32246f1a39e : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620168ca660f4385ac2068f173e282e3 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59cd9fc135d2408c8b830b53c00953e3 : SUCCESS in 10m 25s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f38c59469b714a408a6422cee5b3a1c8 : SUCCESS in 48m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f6850b896f34271b1b868136b566f98 : SUCCESS in 49m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ece9d844a49548c4ab9d46ec1bf788ce : SUCCESS in 48m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1f4f11196f2542fbb222ae1134248677 : SUCCESS in 49m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/85c15bdbcc6b4583a1bb4dcefe0ffdc1 : SUCCESS in 50m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a3c84882be904b54855a34c3bac0b9da : SUCCESS in 30m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fad0726088ba4ad78ff94c17dbd93272 : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e075ee8b80a1d83e2741c0f2667087785e9c6742","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:38:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dd491309e27e51d8219d95353a3aeabc55b8cd51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:41:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c7341d3a024856a8e03a893a4958cb : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87c63f4206e32fd1fd7ff793afc664a567a2deba","revisions":{"87c63f4206e32fd1fd7ff793afc664a567a2deba":{"kind":"REWORK","_number":1,"created":"2020-10-05 10:26:57.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/32/756032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/756032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/756032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/756032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/756032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/756032/1"}}},"commit":{"parents":[{"commit":"175e5fd5877ba062511fd41fb73261faca35b15e","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/175e5fd5877ba062511fd41fb73261faca35b15e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-10-05 10:22:28.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-10-05 10:22:28.000000000","tz":420},"subject":"Remove a TODO comment about saving an action spec","message":"Remove a TODO comment about saving an action spec\n\n* It\u0027s clear now that we don\u0027t have to store an action specification\n  as part of the corresponding action execution object because\n  the notion of action specification itself is specific for a certain\n  type of action. In our case, ad-hoc actions.\n  All changes recently made in the Mistral layers above the engine\n  prove the correctnes of this thought. The comment can be safely\n  deleted.\n\nChange-Id: I45b97b08184c8d5a88bcc537fb5b1e538f105554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87c63f4206e32fd1fd7ff793afc664a567a2deba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87c63f4206e32fd1fd7ff793afc664a567a2deba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
