)]}'
{"id":"openstack%2Fmistral~762307","triplet_id":"openstack%2Fmistral~master~Iaccc059aad4d31a413844bfc252d544a591dfc95","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Iaccc059aad4d31a413844bfc252d544a591dfc95","subject":"Import HTTPStatus instead of http_client","status":"MERGED","created":"2020-11-11 07:44:38.000000000","updated":"2020-11-16 12:56:54.000000000","submitted":"2020-11-16 12:55:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762307-1605531307558-33b2152c","meta_rev_id":"b28b37caed88bc5097611e7ee2bf1c13935a8d84","_number":762307,"virtual_id_number":762307,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-11-16 09:20:45.000000000","_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-16 12:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-11-16 08:18:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-11-16 03:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-11 14:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2020-11-16 03:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 14:53:51.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-11-16 03:13:43.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-11-16 08:18:13.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-11-16 12:55:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b720d537dc1a51dc064227383d8ae98ac93aba11","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-11 07:44:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e75f504ba62122b3ec838e9238275413e05bb1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 10:18:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da749a184efa496b8871ae887d9dfb92 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6df61e8bede450488c1584b0119880e : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69d587ade6642819add7ccb99bb1420 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94987738fb3c41d7aadd892fbfe510f6 : FAILURE in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0161087314b84d7ba27850e7d3b79bef : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21ceb728acd54c5dbf7647f193a9e775 : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdbefac78ba54246a25bffb9ff9f7880 : SUCCESS in 9m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f928eb51a9214cc3b4a26bc8276a4b2a : SUCCESS in 46m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5653797cb48a4904bbb90ca35e10f71e : SUCCESS in 48m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63805163c2c44ec390ac0024948e2c3d : SUCCESS in 49m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1a553ceddd544815bd0c81247d415da7 : SUCCESS in 39m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e4e12eb3870942ecbff63d8289048184 : SUCCESS in 36m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/43ad8fb5d9a54104b5c69df9b2d7b0fb : FAILURE in 30m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/229618bc00a34bddacba640f3d18c32d : SUCCESS in 27m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4d198aaaad404edc8c330772b8326179 : TIMED_OUT in 2h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4da1c292bd6dfffbc823157846e83d24552aa67f","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-11 12:05:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ccbb5e60667ca8d1d0d9edccaf0468dd7301808d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35fe8f17cd464fae83a3c45384ffbe31 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcd34021d2f94b7c9cd1b553b542b512 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed0f84b010e542d39a553fad7c61ac84 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feb14d5a4b6747bba75b8cb7a1274e2a : FAILURE in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3da24daa533474f975ab960e144b3b5 : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48d34d5e614a4d4e8a11cfabd5235f6c : SUCCESS in 8m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec6e79905bc1410396f784b81e1c2087 : SUCCESS in 8m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/728a8f9464f444dcb9f4b7a494d86ff9 : SUCCESS in 42m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3f57ee01796442a8cd748522e7c67eb : SUCCESS in 37m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6f892b05d144a50bdd0a3b34400cb5b : SUCCESS in 1h 04m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/71570a55511047fca4721c16432c36af : SUCCESS in 47m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f6093aa926534423bf690efcc1d109e2 : SUCCESS in 52m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/873ba43cce134c73855894488a4a9128 : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a58ce7426c35477baba06fed4aaaf81b : SUCCESS in 26m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a6389d72c5246fb8dd3b2ddfc9a98a6 : TIMED_OUT in 2h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e4fef58952f09abba83737d4eb898910223ae89","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-11 14:53:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3304aad8a834edbd8056df0040c0f242fe17bd00","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-11-16 03:13:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"284e504a2623990e24b36f931d4ea49ae1972f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 03:13:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4761ad7c9069b7badf4f2ceec73710c7b438f9bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 04:08:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c876bc27ada24b92a27951467904d85c : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b13bcee85ccc43b38e21b3b8e201b64e : SUCCESS in 10m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22d8dd8c3ad451eb5d9ce032dd79b21 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b22160e310246d19000907c05d00955 : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/357552e63e5f46dfb8319b7431e2e245 : SUCCESS in 8m 25s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fa57b7cb83ed4131878f314d8457e502 : SUCCESS in 40m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87f44eca79c7428ca7fb4b07ab0d4e11 : SUCCESS in 39m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e51fc77dfd744639b63c388603aba00d : SUCCESS in 46m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0aefbcc6203842638fed844c8d741612 : SUCCESS in 38m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cc3256dd68764802a6b6b2105cb9867a : SUCCESS in 53m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bbe037febcee48eebcf01f347f204b80 : FAILURE in 29m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f4c1a9fd28c54b66af85d5a76c0e75e4 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cbc3a82b080f617e2eec8c202629bc68f7065886","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-16 05:58:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"481a19d24c2bfc2298d14d46afc06d8964640467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 08:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a24acfc7b145a7b6421c223daea6cb : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40f84ce36b8046fdaeb4874a0b1c2790 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd7c707f7f9422998583a2058d58f23 : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66f844aa586d410381d2b975e3d5cc60 : FAILURE in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b474fb69fd68454fb23ed4264923a3b2 : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/995c0c75f45b4fa992c30efce9e76ba8 : SUCCESS in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b250925b09654f26877df229bae765b3 : SUCCESS in 9m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7be8263c2ab34adc9925bf34ac98fbba : SUCCESS in 47m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/743b5eff97be4bbcb055c0938df1921e : SUCCESS in 1h 01m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0053e01810407ca7e80fbd7bc15ddb : SUCCESS in 43m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e81a7f0f09db4b16b02ee80d56b6a33d : SUCCESS in 43m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/30566b8995664d62b98b7e985b761bd1 : SUCCESS in 48m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/044619e6f817432aa92671bda0f73812 : FAILURE in 28m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8155791db492428b9d189764e46fea7f : SUCCESS in 26m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/31bc8d2f1f154759b9032e45a4402e2d : TIMED_OUT in 2h 04m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8a3b3d74e82c8852d3a234f8eefe52486778e6c","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-11-16 08:18:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aed0fad8406670ebd0fba841293172849316e76e","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-16 09:20:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"29fc76327058542aa005772d9d0cd34bd22358a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 11:34:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab9e22f49c94460affb7e73dddc2c5b : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e8aeb6fc02a4cd7873f2f6856da0f24 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9285d220ea934ec185c039e7ca40297c : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46840b5a469047d98fb89bc6c9a7606f : FAILURE in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd71496d0b3548cb9ca64b93f283715f : SUCCESS in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4c335ee49f0469fa09c78224408ad7d : SUCCESS in 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de32776cb064e639971e95ded5d7231 : SUCCESS in 9m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9a4f43f0be344db3962528c8ce7afc80 : SUCCESS in 50m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/034a977f03e046f8908251399510fb71 : SUCCESS in 45m 22s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9d85e8f1de64b0f8e05e4f0483a21e2 : SUCCESS in 52m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cfedaaf6be13493f8b4501cc1fc2e9d1 : SUCCESS in 58m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/60458a3047bf433e90e85ac104a4fba9 : SUCCESS in 46m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/854d16fbc5324b2781872565a25c18b0 : SUCCESS in 28m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/97f8bc359ebb4a9aad12a7f96ed998c5 : SUCCESS in 27m 40s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bc31bd99432f468296cedb518dfcb1ed : TIMED_OUT in 2h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0ed7124f14cf13f1f431174719e20a50ee177ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 11:35:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6e59c723d039f95897257e8bf3a3813d0ed5d53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 12:55:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a9fbb02387d71a18b47baff824a3c9bcc274911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 12:55:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9110309431824061a478b49dbe191340 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5466e3ca36e3434c836f4ae657338e78 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e62d8ce44f944ecb16d9f2ccf927431 : SUCCESS in 19m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e4e7e40b864331ab102fc332267c48 : SUCCESS in 27m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18f1ad41e1d9494fa44a08564b975d5c : SUCCESS in 12m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2557749093974300a90c7006873fd791 : SUCCESS in 49m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f72d553582da4bdd8d03ecee682c5989 : SUCCESS in 56m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/364b0295454a48b3a98f800a8cf65327 : SUCCESS in 53m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1b00805324384a57be04292cc56cf482 : SUCCESS in 50m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6358bb08fa3f420cab8eb71bf5dd2a7d : SUCCESS in 1h 10m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/60be48e2fe3c480b97dea4ac1e5b9e28 : SUCCESS in 29m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/20476ec6b39b4f33af73b1bde1b4696f : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b28b37caed88bc5097611e7ee2bf1c13935a8d84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 12:56:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09124a8d6beb4fd59ff69affcb8ca03b : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d3509e28ce4f5b06d000d43a1862f68300290d3","revisions":{"0d3509e28ce4f5b06d000d43a1862f68300290d3":{"kind":"REWORK","_number":1,"created":"2020-11-11 07:44:38.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/07/762307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/07/762307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/762307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/762307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/762307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/07/762307/1"}}},"commit":{"parents":[{"commit":"2cd445587368755e7846061728b10ead86790c60","subject":"[doc]Fix invalid url link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cd445587368755e7846061728b10ead86790c60"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 07:41:51.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 07:44:33.000000000","tz":480},"subject":"Import HTTPStatus instead of http_client","message":"Import HTTPStatus instead of http_client\n\nWe don\u0027t need an HTTP client for this, we only\nneed status codes.  Just load that instead.\n\nhttps://docs.python.org/3/library/http.html\n\nChange-Id: Iaccc059aad4d31a413844bfc252d544a591dfc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0d3509e28ce4f5b06d000d43a1862f68300290d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0d3509e28ce4f5b06d000d43a1862f68300290d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
