)]}'
{"id":"openstack%2Fmistral~762954","triplet_id":"openstack%2Fmistral~master~I8fac507bbe0da44248ed0fa0ba66771c2977831c","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I8fac507bbe0da44248ed0fa0ba66771c2977831c","subject":"Update TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2020-11-17 06:02:41.000000000","updated":"2020-11-17 10:34:04.000000000","submitted":"2020-11-17 10:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762954-1605609080322-33b2152c","meta_rev_id":"a621ed625089c6c5d6864ab67544e9b35e6f1824","_number":762954,"virtual_id_number":762954,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 10:31:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-11-17 09:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-17 08:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-11-17 09:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 08:28:28.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-11-17 09:27:21.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-11-17 10:31:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78420a3d4dc5bb75624886b6d552c48f05be8417","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-17 06:02:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06bd93fd24f1aa2f004de19dbf37d93b982d324b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 08:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27700eae75fb4235854f464bcacfb997 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3c76e140deb4c6eaef42e05cadaf6f4 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baee826fee67408d9f66c7f74985f2ae : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/682a709566b44eb5b730a3e06d9122c1 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564a58e362a14f4480fbf25f67dec3ce : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db4bdc7cc23f44b8acf6fd7327f11cd3 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a04b722072e4725ad18822c835963f1 : SUCCESS in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7103b9ad6c3644eeacb1dddcb0b6d618 : SUCCESS in 8m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0cb1fcdbf906468b82dd5416a79ba3aa : SUCCESS in 42m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c91337e5f8494606899fa3b2927da4e8 : SUCCESS in 43m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34614117cdd84ff890ee976a65dba73a : SUCCESS in 1h 03m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0899787edebe43ecb7547dd2c7843d58 : SUCCESS in 48m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b127723dd949414bb29f53765bdabf3d : SUCCESS in 1h 06m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cd90950d3f1e42b09e6381f63668f897 : SUCCESS in 33m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e90cc59697cc4086abb44db3e7de0478 : SUCCESS in 28m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/56fbf398a417470fa58da1f9dd3a21d8 : TIMED_OUT in 2h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef8af4bfdba12042f81987c758daa56b03a2bcbd","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-17 08:28:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a564ee7c6f2ab7ce8317a297941c415589649f5a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-11-17 09:27:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b72080c05c3731d11b88c6e7567074e1d52b3c8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:27:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4577b65a5be17abcd4be9ab329d9348040336a64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:31:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa40cfa9e2ca785cd16e5d1e55a79c41f10faac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:31:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c862e8efaac44f78f11ec3cf789f1c1 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a09c1b0deb6441e89daabc4edd119b3e : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1751750e38674e96b472138cc2355d08 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fcb7ac301f482dbe11c2aabd7f3a22 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d3e9ae80dea49e693e6594312e908b4 : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7efa07062074169aed29e7b7e3dc064 : SUCCESS in 8m 39s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6b661c13ee394478aa5f50a7c6f64d30 : SUCCESS in 56m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4282fb69e8cf4f4e8e526e5485f9556d : SUCCESS in 51m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0102f47750b46f98d73e14c084003f1 : SUCCESS in 49m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/220ef9af26324a75ba6637c1e0e46af1 : SUCCESS in 43m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/880b6690753046e48d111725c8a944ca : SUCCESS in 39m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d01671ec8e3648019423d6184e94a746 : SUCCESS in 33m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9f2dfceb9de544b49913aeae7a4e0abd : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a621ed625089c6c5d6864ab67544e9b35e6f1824","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344d85f9e6cc4eb4b45e518e19d8f324 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ba1d877b2d347e88a26b5a5eb9df3f8 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bce0747663df1e1d7fdc05147fe6361cccf6ce16","revisions":{"bce0747663df1e1d7fdc05147fe6361cccf6ce16":{"kind":"REWORK","_number":1,"created":"2020-11-17 06:02:41.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/54/762954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/54/762954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/54/762954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/54/762954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/54/762954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/54/762954/1"}}},"commit":{"parents":[{"commit":"0d3509e28ce4f5b06d000d43a1862f68300290d3","subject":"Import HTTPStatus instead of http_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0d3509e28ce4f5b06d000d43a1862f68300290d3"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-17 05:57:54.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-17 05:57:54.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n\nChange-Id: I8fac507bbe0da44248ed0fa0ba66771c2977831c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bce0747663df1e1d7fdc05147fe6361cccf6ce16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bce0747663df1e1d7fdc05147fe6361cccf6ce16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
