)]}'
{"id":"openstack%2Fmistral~764896","triplet_id":"openstack%2Fmistral~master~Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc","project":"openstack/mistral","branch":"master","topic":"polish_dynamic_actions","hashtags":[],"change_id":"Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc","subject":"Fix dynamic actions further","status":"MERGED","created":"2020-12-01 05:19:13.000000000","updated":"2020-12-07 09:46:12.000000000","submitted":"2020-12-07 09:44:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"764896-polish_dynamic_actions","meta_rev_id":"19e2f3b84097ee6496bc7ec7eb9d4e830a255582","_number":764896,"virtual_id_number":764896,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-07 09:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-12-07 06:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-07 08:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-07 08:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 07:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 10:20:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-12-03 10:20:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-12-03 10:20:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-12-03 10:20:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-12-07 06:22:48.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"}],"messages":[{"id":"29f144b43ed98d9f031a1b3cbe4a3dd52c11eff7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-01 05:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b07c3c6473f3751c8d247f6d8c59574c46e8d464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 07:28:03.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218e24bb36ad47c69904fa34b114516d : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/599af787a7894b56a184dac25f9c645c : FAILURE in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a2f07552e8d456a84d335e6c7e27b5e : FAILURE in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcfbc482517a48b5977b7e3dc2f81c7b : FAILURE in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a675573fb8404a889d0f80d225046c6e : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77d5c160934f441fbbcb8c17d74d8f1f : FAILURE in 7m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58f17d8f8dde4ce18834faaf036981c3 : FAILURE in 8m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/55f9834787a74a278629f06674739328 : SUCCESS in 49m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b9b097369f94d768e3fc00a91c4efbc : SUCCESS in 52m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc3c7ff4480748c69bcdb21cff80785b : SUCCESS in 50m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fa57a286c77f4bf0bc488e9297ba0b92 : SUCCESS in 58m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c2e93144dde44aa88a369416f13fdccd : SUCCESS in 50m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/57913b2ffbe7482aaa8a132647432ea3 : FAILURE in 28m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/55a6d7d3c9cc458a88180cd1669f20f5 : FAILURE in 26m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e80c5e7cbff543c4a7e61d993872197c : TIMED_OUT in 2h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"083ab7454c8b94711dbe785efc652eaec2740392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-01 09:19:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4f6cab9d238e7ac9ccb6f6f46418a3438300466","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-01 09:25:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"37b653b5537bdee11af711f83c29f7e28c1402e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 11:35:31.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd16fefa664425694ff61f8a98012e9 : FAILURE in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6f47272fc647f38aed8fa797bfd6c4 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/811b91f3962e48d78a92d0b9c5f000d1 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3efe79ea2dd47fcba9c229e64823196 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9abc8922db524e9e9a2504c595f1691f : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98bfd364f5aa439581631fc75f492c75 : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b1a719665304a66858c36ce261bcd16 : SUCCESS in 8m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4e93fafeb7a84a0d99afdceabb6ae234 : SUCCESS in 54m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56a4fc4f7c4f4551ba4e5b7c093b82d8 : SUCCESS in 50m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6255486a870b43a3b47dfc817cbd8ddf : SUCCESS in 49m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bbdfa1823bda4c02b000382465ac98f5 : SUCCESS in 59m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/aa14d75432314acfafe4f6e638bf96c9 : SUCCESS in 49m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/00dc14ce6b6b4a7caa4acb904bcd510d : FAILURE in 28m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/619cc2fd8d804a59a04b82cb4b9a2636 : FAILURE in 26m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb7bf0f068ff4d36b54ac395951136a8 : TIMED_OUT in 2h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f0bcb4b61d5993287b06ec6644e045b48b0f72fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-02 04:38:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f915e3d4aea45b42e9e507723e715c95d4ba7b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:47:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfecfd702bc493c9bc6192dd295ef09 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363220ea53b3403483e3ba42517bbfb2 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dbec8891b704834a431ab26071fab6b : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebf8b4b2ca314682bb3b3e9b89881cc2 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a3e77e110b94773a77e5c63289aadd1 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be83d97b06f54c968922b8567c27f6b0 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91ca762475814e49beb6584ea90478e8 : SUCCESS in 8m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/11707bd37129451d8d9c49302336b32b : SUCCESS in 54m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/389d1ca991c4473daa695b5270b144d5 : SUCCESS in 42m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bbf349964e84acc93891f1914235c0d : SUCCESS in 46m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d8eb9fa590734ed08d0c3f6f4081979f : SUCCESS in 56m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/91184b1f2ebb4d1cacaac338f46e0d6d : SUCCESS in 59m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9295ed7f65ba4e75b4ab4ae2c8e4aef2 : SUCCESS in 32m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/20666a43bd3f4c0a82ba22715ac5dcb2 : SUCCESS in 26m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5fbcab5af18b476093e9e78bb14f7842 : TIMED_OUT in 2h 04m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"30ca9fbb5832fece0f612dacad19e76fcc5f0f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-03 08:06:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e2fea1896334da67cfb4ce4bd1427217a24634f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-03 08:24:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7dc7b243663966e9e26c5ac8ab52b533047c9935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-03 08:25:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1fa23b94f62e831af9959a9806f7c3adde292b7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-03 09:17:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"94eab62c9f86514a9becd35ec998498560892cc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-03 09:36:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dc8975d2dd12a97966fe384fc758b2e4cd9f12c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-03 09:51:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e2c671eb46336222a49d815e231182e4bcd70f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 12:02:11.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af884a475bb46e4aa27e481f08ccf82 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f2a43a5689474299d53e3cdc79701b : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8f4f9c7fe5a4fd8a22db12a4c8cb734 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b3290ab386d4c81b090f0a6c94d0984 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c86d76a077423cabdd29b65ee6d70f : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81cbbf30a4b84de486d2882bfedefc36 : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/339985161ca24362a6f5c52a558c472a : SUCCESS in 8m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6288d77515d9409c8436a6e9d54f7cfb : FAILURE in 38m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/323e1a74e7c7432babe0152092d9915b : FAILURE in 43m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8267245dabe14bb7a9c26a754f36ccec : FAILURE in 47m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c59d520bd204212986367627d8b5c0e : FAILURE in 31m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/88c6f730eb4e44aeac1cea741379ac0b : FAILURE in 37m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/03be12950cbd4b5488e29b78e14cf6e3 : SUCCESS in 29m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/863040ae703640b5884ff2ca898524d6 : SUCCESS in 27m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6de7ed22cc364120af305127e3c167e6 : TIMED_OUT in 2h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"941f77eccf40473d47bf453d7cff3219b12b40fe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-04 06:39:09.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"1aa47758c0f172154033b2872f5884576aae2883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 09:47:31.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da4ec0da70a4ee28b3ccbe451af17fc : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb8bf64a1a14fddadd3d771a28f3f6e : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e6114c6781f4b618172bd1acb96205d : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f334223c46f54618b92245e9fcaf4caf : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39be0b0e5b040c48a73afcd961aaf82 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87dac252accb4dc1988d4c4a9e9f9488 : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8dc513fa986430cb9f4339d2aab7713 : SUCCESS in 9m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fae8b88424aa4719a9b2dd25cdfc9cf6 : FAILURE in 38m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4695cb401d284b11ab1bff4f303f73fa : FAILURE in 40m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94e8fea5599a48a89952f4a2faab7bb7 : FAILURE in 48m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6fd1636ee34c4eff88c3419035c6aae4 : FAILURE in 47m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2e182b947c4f4f80aa5adb3e111082a2 : FAILURE in 46m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e78bc39b29e4400081a4afbef7e2c7d6 : SUCCESS in 29m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/254e7c0102d0413dbb140cd9a4e2f065 : SUCCESS in 26m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2f2cc07acaa740cbb7382d7d170e3e8f : TIMED_OUT in 2h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4e1fa4c99e06a53bdce613c8c1efefe6756a5228","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-05 12:31:36.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"1bc960bc3ac95d8ffaa00cd3d60ceef801d0f3b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 14:38:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9caf89918684da5bc2df288e6ec5927 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df7545561dc14d88bdd36b2c3454e17c : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/200a34b8387549ddb9c1e6a2213be754 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5ea7eea376d4a95aac88cc1ec80f673 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c4a01fd5f44e17afa7417890e4f3ec : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0477db577cfd42c0af728fc256c2d7f9 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57499eaab76428880e049e3bfab84a7 : SUCCESS in 9m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ba270afa2ba14959958580a8163985a7 : SUCCESS in 43m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/472240d199f34ea8892dc9ad0539da50 : SUCCESS in 48m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30fc362b1ad24613a921064ff80ac5d7 : SUCCESS in 43m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/86995da229f045d099099193e72ac5f6 : SUCCESS in 48m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/05c094c1df184d76bbfd1496f2866b99 : SUCCESS in 46m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/18262a19621343b78ff4392ae2047bb6 : SUCCESS in 31m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5dfe2fec020d46dcbdd5b2dee8089284 : SUCCESS in 29m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a9b25efe71734cf186c7c545d6137f43 : TIMED_OUT in 2h 04m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fa28c0354f0da29c2c087e2d64e1aae12eed4968","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-12-07 06:31:25.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nI love the code deletion 😊","accounts_in_message":[],"_revision_number":10},{"id":"f04f3a0638fb4a3c82206bca27a053b234b3d679","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-12-07 06:31:33.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"767e87672c95ef7dd6c5c114630efe56d2121c27","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-12-07 08:38:45.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"e1e0f80f45a48ac8d0d68e5a7b6e2f2f90a6e974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 08:39:02.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d7582a6633f0e28f3247c4fc92d933168c7a864b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 09:44:50.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89080866035d492ebb102c752b9e3615 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/323caf8be4914fabb8ad4ce82a8822e2 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b3cf0f049a0482ca82440ade1545004 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be1108647b54b9ba3530fa507c733ac : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ec08ed01e444b3da688d2c59143e673 : SUCCESS in 8m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/592a01c7bbf14518a0962b277c3e2e48 : SUCCESS in 40m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11ebe0fea7164a58b16689ef23dd1143 : SUCCESS in 53m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc3bb2bd05334a2dac73a36f0123dcf5 : SUCCESS in 46m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a39f32b697e24cc6b82b1605b814014f : SUCCESS in 56m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f91ec5f625ef4e14bb0b96a54e52a1ad : SUCCESS in 49m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/05bd8981a70242acb4286f3b455a9eaa : SUCCESS in 28m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5f939acb66ee4753a9f6431f75bafeaf : SUCCESS in 27m 26s","accounts_in_message":[],"_revision_number":10},{"id":"8c7989363af489d4074936a417ec683b5fa992b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 09:44:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"19e2f3b84097ee6496bc7ec7eb9d4e830a255582","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 09:46:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803683c0483a460eb74b56db256cc5dc : SUCCESS in 56s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","revisions":{"2b9b0bb43f09ea68403de2cb2fa5635a10e9d535":{"kind":"REWORK","_number":1,"created":"2020-12-01 05:19:13.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/1"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-01 05:18:51.000000000","tz":420},"subject":"WIP: fixing dynamic actions further","message":"WIP: fixing dynamic actions further\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b9b0bb43f09ea68403de2cb2fa5635a10e9d535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b9b0bb43f09ea68403de2cb2fa5635a10e9d535"}]},"branch":"refs/heads/master"},"a2969bf8ef4912e4bdeb902d300dfaa0127a0f92":{"kind":"REWORK","_number":2,"created":"2020-12-01 09:19:49.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/2"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-01 09:19:32.000000000","tz":420},"subject":"WIP: fixing dynamic actions further","message":"WIP: fixing dynamic actions further\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a2969bf8ef4912e4bdeb902d300dfaa0127a0f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a2969bf8ef4912e4bdeb902d300dfaa0127a0f92"}]},"branch":"refs/heads/master"},"2e81d216646f8912df28bab88dac09da92bf7e39":{"kind":"REWORK","_number":3,"created":"2020-12-01 09:25:11.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/3"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-01 09:24:58.000000000","tz":420},"subject":"WIP: fixing dynamic actions further","message":"WIP: fixing dynamic actions further\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2e81d216646f8912df28bab88dac09da92bf7e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2e81d216646f8912df28bab88dac09da92bf7e39"}]},"branch":"refs/heads/master"},"deb4aa5e305906645c0c8fe375300ad641a11721":{"kind":"REWORK","_number":4,"created":"2020-12-02 04:38:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/4"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-02 04:38:03.000000000","tz":420},"subject":"WIP: fixing dynamic actions further","message":"WIP: fixing dynamic actions further\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/deb4aa5e305906645c0c8fe375300ad641a11721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/deb4aa5e305906645c0c8fe375300ad641a11721"}]},"branch":"refs/heads/master"},"c54ea70f0b78a0d432e0f1df556169e49eefacbb":{"kind":"REWORK","_number":5,"created":"2020-12-03 08:06:51.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/5"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-03 07:59:37.000000000","tz":420},"subject":"Fix dynamic actions further","message":"Fix dynamic actions further\n\n* Reworked /code_sources and /dynamic_actions API endpoints to\n  simplify them. For now they don\u0027t work with multiple objects and\n  they are consistent with other endpoints. If needed, we\u0027ll add\n  support for multiple objects (i.e. adding multiple dynamic actions\n  with a single request) later in a backwards compatible manner.\n* Simplified unit tests.\n* Got rid of services/*.py modules since they didn\u0027t do anything\n  useful. They were just wrappers around DB API calls.\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c54ea70f0b78a0d432e0f1df556169e49eefacbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c54ea70f0b78a0d432e0f1df556169e49eefacbb"}]},"branch":"refs/heads/master"},"60aa8523422666e52cdb79522989fc121cf52b75":{"kind":"REWORK","_number":6,"created":"2020-12-03 08:24:05.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/6"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-03 08:23:52.000000000","tz":420},"subject":"Fix dynamic actions further","message":"Fix dynamic actions further\n\n* Reworked /code_sources and /dynamic_actions API endpoints to\n  simplify them. For now they don\u0027t work with multiple objects and\n  they are consistent with other endpoints. If needed, we\u0027ll add\n  support for multiple objects (i.e. adding multiple dynamic actions\n  with a single request) later in a backwards compatible manner.\n* Simplified unit tests.\n* Got rid of services/*.py modules since they didn\u0027t do anything\n  useful. They were just wrappers around DB API calls.\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60aa8523422666e52cdb79522989fc121cf52b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60aa8523422666e52cdb79522989fc121cf52b75"}]},"branch":"refs/heads/master"},"a20cd21fe9142a3bd4c87bd0b6ab1283f54245c7":{"kind":"REWORK","_number":7,"created":"2020-12-03 08:25:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/7"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-03 08:25:46.000000000","tz":420},"subject":"Fix dynamic actions further","message":"Fix dynamic actions further\n\n* Reworked /code_sources and /dynamic_actions API endpoints to\n  simplify them. For now they don\u0027t work with multiple objects and\n  they are consistent with other endpoints. If needed, we\u0027ll add\n  support for multiple objects (i.e. adding multiple dynamic actions\n  with a single request) later in a backwards compatible manner.\n* Simplified unit tests.\n* Got rid of services/*.py modules since they didn\u0027t do anything\n  useful. They were just wrappers around DB API calls.\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a20cd21fe9142a3bd4c87bd0b6ab1283f54245c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a20cd21fe9142a3bd4c87bd0b6ab1283f54245c7"}]},"branch":"refs/heads/master"},"82dc271ed0f12fd84b7c0a263bad14dfab03ec43":{"kind":"REWORK","_number":8,"created":"2020-12-03 09:17:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/8"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-03 09:16:48.000000000","tz":420},"subject":"Fix dynamic actions further","message":"Fix dynamic actions further\n\n* Reworked /code_sources and /dynamic_actions API endpoints to\n  simplify them. For now they don\u0027t work with multiple objects and\n  they are consistent with other endpoints. If needed, we\u0027ll add\n  support for multiple objects (i.e. adding multiple dynamic actions\n  with a single request) later in a backwards compatible manner.\n* Simplified unit tests.\n* Got rid of services/*.py modules since they didn\u0027t do anything\n  useful. They were just wrappers around DB API calls.\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82dc271ed0f12fd84b7c0a263bad14dfab03ec43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82dc271ed0f12fd84b7c0a263bad14dfab03ec43"}]},"branch":"refs/heads/master"},"f979040b160a9014da2ff303f82ad7bf53512f13":{"kind":"REWORK","_number":9,"created":"2020-12-03 09:36:55.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/9"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-03 09:36:41.000000000","tz":420},"subject":"Fix dynamic actions further","message":"Fix dynamic actions further\n\n* Reworked /code_sources and /dynamic_actions API endpoints to\n  simplify them. For now they don\u0027t work with multiple objects and\n  they are consistent with other endpoints. If needed, we\u0027ll add\n  support for multiple objects (i.e. adding multiple dynamic actions\n  with a single request) later in a backwards compatible manner.\n* Simplified unit tests.\n* Got rid of services/*.py modules since they didn\u0027t do anything\n  useful. They were just wrappers around DB API calls.\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f979040b160a9014da2ff303f82ad7bf53512f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f979040b160a9014da2ff303f82ad7bf53512f13"}]},"branch":"refs/heads/master"},"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704":{"kind":"REWORK","_number":10,"created":"2020-12-03 09:51:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/764896/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/764896/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/764896/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/764896/10"}}},"commit":{"parents":[{"commit":"f78f33507e55a6c989125840d57c484c071d9ace","subject":"Code improvements after the dynamic actions patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f78f33507e55a6c989125840d57c484c071d9ace"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-11-24 10:03:05.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-03 09:51:25.000000000","tz":420},"subject":"Fix dynamic actions further","message":"Fix dynamic actions further\n\n* Reworked /code_sources and /dynamic_actions API endpoints to\n  simplify them. For now they don\u0027t work with multiple objects and\n  they are consistent with other endpoints. If needed, we\u0027ll add\n  support for multiple objects (i.e. adding multiple dynamic actions\n  with a single request) later in a backwards compatible manner.\n* Simplified unit tests.\n* Got rid of services/*.py modules since they didn\u0027t do anything\n  useful. They were just wrappers around DB API calls.\n\nChange-Id: Ib5a53f1f1a185f0395ffae1ab0c401633fcdd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
