)]}'
{"id":"openstack%2Fmistral~766877","triplet_id":"openstack%2Fmistral~master~Idfc2ae6db344cc788ddf7f0b3883b678af213621","project":"openstack/mistral","branch":"master","topic":"fix_low_constraints","hashtags":[],"change_id":"Idfc2ae6db344cc788ddf7f0b3883b678af213621","subject":"Fix constraints","status":"MERGED","created":"2020-12-14 04:17:55.000000000","updated":"2020-12-14 11:11:50.000000000","submitted":"2020-12-14 11:10:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766877-fix_low_constraints","meta_rev_id":"80c928eb08f9ddca46358fe9b747030e7bac821a","_number":766877,"virtual_id_number":766877,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-14 11:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-14 09:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-12-14 10:11:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 06:47:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-12-14 06:47:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-12-14 06:47:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-12-14 06:47:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-12-14 07:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75d4eacee2399ff7f00103d009274a7d4aacf8c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 04:17:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e3e7a2f44dc7478e8d0240d091fa61342ccc20e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 04:34:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b615e1a6789a128fb7ed0cb8a21d4dc56ea4c313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 04:50:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd8ed25ac735e90857deacd0fadc4da495d6d8f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 05:02:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d25bccbb4b523c0cd188f059b9a9ba358a23759f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 05:22:02.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"36b9d9a1508f4849271193f068d102b297413d4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 05:24:40.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"d704a7cf0b55d2a8448eef7964bb6d40aad5f722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 07:31:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609d29829ef24673b9274499386a23fa : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bd76266df274ec486fdb2ffef179c4c : SUCCESS in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e5c26d098ee4452a26e6b1fc3871d1e : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/560054cb93024a1db128c660f20be742 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57601261cffe47c19b068e1abeccb4d5 : SUCCESS in 9m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8cc31e9f2d3c4b2ea5198868d537947b : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c053f6627c5417c97ae9cf64cfb9b66 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9de771133c5246abbcac448870018850 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9d6ddb4a2de4bc982a88d37efcffd86 : SUCCESS in 9m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8bb43ea5796742f9868e691d5c220903 : SUCCESS in 42m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ae32a7d0b7e40899e4cee27fc8e1c63 : SUCCESS in 1h 08m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ae8199994a045339a1cd827a4a89ee4 : SUCCESS in 55m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0be75ac6bb9b416aa328c56498a1d060 : SUCCESS in 1h 09m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b30763adbb8946718d3694eb1184db07 : SUCCESS in 49m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e4a6df21d1bc4eafa2f217c14160c823 : SUCCESS in 28m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1fe1264007c246daa3f62af5e14783ed : SUCCESS in 28m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fe01695e2266461aaf485ab9c94f7660 : TIMED_OUT in 2h 03m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8ca44c2f3e41c654c616bdeeffbb832006066523","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-12-14 09:34:21.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4b17e02f5723b1d783e19eda7cccb4157df4d66d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 10:11:20.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9aa2434c23d6877cc147c83323435efd64cbacc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 10:11:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4084f25b8698af556ea7117ee11e3ab1115a7660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 11:10:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4b85d9eb2244e0bd521f3b57e70081 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd7c6225b4c94289962d5ab620a272d8 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/110086d629d3495b9e20b6ecea3f0d35 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cff66050304015a7ddf95736dad8c4 : SUCCESS in 7m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c921af1e28094466b17b5327f5426ab4 : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df42616d3fd74426be0f112dfea40b3a : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/240c0ec185b04906a9e09d7ab4c8867f : SUCCESS in 9m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7dca39686b3d4eaca219ee4c9c4c6926 : SUCCESS in 42m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d2167ede8da4d8d8e895af9c9baf319 : SUCCESS in 51m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cddd60cfe3cf4312b5ca58bc35147661 : SUCCESS in 51m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dbeaf00102454ea3bf5cae67974fe4cf : SUCCESS in 52m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/adfb7c8df63a41669a8c84e0a5bbdcb7 : SUCCESS in 50m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aaa07c0b33744e17880cf2ee90ad1c92 : SUCCESS in 29m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6c9337334eb645d095641cd7305385b3 : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":6},{"id":"46ed3899ee039fd874924772b1eff9deaa715f4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 11:10:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"80c928eb08f9ddca46358fe9b747030e7bac821a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 11:11:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ce5af8410947c59d1eba6d7d9c429a : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3249bdddf10c4d738d0ddb2510539c10 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1c44a2b1db1fd8b02f4a117f7a5a62560e26375e","revisions":{"0199c992794c08d404f5d22abb488d708f5a19c3":{"kind":"REWORK","_number":1,"created":"2020-12-14 04:17:55.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/77/766877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/766877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/766877/1"}}},"commit":{"parents":[{"commit":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","subject":"Fix dynamic actions further","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"subject":"WIP: fix low constraints","message":"WIP: fix low constraints\n\nChange-Id: Idfc2ae6db344cc788ddf7f0b3883b678af213621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0199c992794c08d404f5d22abb488d708f5a19c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0199c992794c08d404f5d22abb488d708f5a19c3"}]},"branch":"refs/heads/master"},"ddf0a1ece998a71722ac80739fe3990319689fa1":{"kind":"REWORK","_number":2,"created":"2020-12-14 04:34:41.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/77/766877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/766877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/766877/2"}}},"commit":{"parents":[{"commit":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","subject":"Fix dynamic actions further","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:34:20.000000000","tz":420},"subject":"WIP: fixing constraints","message":"WIP: fixing constraints\n\nChange-Id: Idfc2ae6db344cc788ddf7f0b3883b678af213621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf0a1ece998a71722ac80739fe3990319689fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ddf0a1ece998a71722ac80739fe3990319689fa1"}]},"branch":"refs/heads/master"},"3d44708e45655c6f5ee607a1e1930134bf817804":{"kind":"REWORK","_number":3,"created":"2020-12-14 04:50:05.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/77/766877/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/766877/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/766877/3"}}},"commit":{"parents":[{"commit":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","subject":"Fix dynamic actions further","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:49:50.000000000","tz":420},"subject":"WIP: fixing constraints","message":"WIP: fixing constraints\n\nChange-Id: Idfc2ae6db344cc788ddf7f0b3883b678af213621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d44708e45655c6f5ee607a1e1930134bf817804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3d44708e45655c6f5ee607a1e1930134bf817804"}]},"branch":"refs/heads/master"},"a5b99b0ab95fb40eb7d3d702a4b17100e51c9189":{"kind":"REWORK","_number":4,"created":"2020-12-14 05:02:48.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/77/766877/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/766877/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/766877/4"}}},"commit":{"parents":[{"commit":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","subject":"Fix dynamic actions further","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 05:02:31.000000000","tz":420},"subject":"WIP: fixing constraints","message":"WIP: fixing constraints\n\nChange-Id: Idfc2ae6db344cc788ddf7f0b3883b678af213621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a5b99b0ab95fb40eb7d3d702a4b17100e51c9189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a5b99b0ab95fb40eb7d3d702a4b17100e51c9189"}]},"branch":"refs/heads/master"},"ae1461fa427a61794f85b80964ed0746fa172e49":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-12-14 05:22:02.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/77/766877/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/766877/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/766877/5"}}},"commit":{"parents":[{"commit":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","subject":"Fix dynamic actions further","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 05:16:53.000000000","tz":420},"subject":"Fix constraints","message":"Fix constraints\n\n* Due to the recent PIP version update requirements checks are\n  now more rigorous. This patch fixes another issue that originally\n  started with incompatibility in the \"mistral-lib -\u003e eventlet\" and\n  the \"oslo.service -\u003e eventlet\" dependency but then went on to having\n  to bump up versions of \"oslo.utils\" and \"tempest\" to avoid conflicts.\n\nChange-Id: Idfc2ae6db344cc788ddf7f0b3883b678af213621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae1461fa427a61794f85b80964ed0746fa172e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae1461fa427a61794f85b80964ed0746fa172e49"}]},"branch":"refs/heads/master"},"1c44a2b1db1fd8b02f4a117f7a5a62560e26375e":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-12-14 05:24:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/77/766877/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/766877/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/766877/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/766877/6"}}},"commit":{"parents":[{"commit":"a73fe5b8a3cbf13ceab26319250d3ae2d21d5704","subject":"Fix dynamic actions further","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a73fe5b8a3cbf13ceab26319250d3ae2d21d5704"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 04:17:34.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 05:23:08.000000000","tz":420},"subject":"Fix constraints","message":"Fix constraints\n\n* Due to the recent PIP version update requirements checks are\n  now more rigorous. This patch fixes another issue that originally\n  started with an incompatibility in the \"mistral-lib -\u003e eventlet\" and\n  the \"oslo.service -\u003e eventlet\" dependencies, so the version of\n  \"oslo.service\" had to be bumped up. But then it also went on to having\n  to bump up versions of \"oslo.utils\" and \"tempest\" to avoid conflicts.\n\nChange-Id: Idfc2ae6db344cc788ddf7f0b3883b678af213621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c44a2b1db1fd8b02f4a117f7a5a62560e26375e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c44a2b1db1fd8b02f4a117f7a5a62560e26375e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
