)]}'
{"id":"openstack%2Fmistral~766906","triplet_id":"openstack%2Fmistral~master~I065a83788bb051c2505ae6ad985b649ecc191f23","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I065a83788bb051c2505ae6ad985b649ecc191f23","subject":"Update doc8 version","status":"MERGED","created":"2020-12-14 09:24:38.000000000","updated":"2020-12-16 04:53:33.000000000","submitted":"2020-12-16 04:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766906","meta_rev_id":"ea7868cf43da3a7c2dc85f1c1909b502d8c9a718","_number":766906,"virtual_id_number":766906,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 04:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-12-16 03:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-12-16 03:47:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 11:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 19:10:04.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"CC"},{"updated":"2020-12-16 03:47:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"db0832218e8edd2dbe0438c0ef15fc7bf552b464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-14 09:24:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa467cd92c0d4c2eb588db5b0714159767ae36da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 11:34:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252ff766d60b4f569aca70c38d3bca29 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bd6a4b1b9064d1ba755f83e2ca74772 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27953fd498634233aad595fea98b19ec : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78126e1526cb449381f5edd17297b8c7 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/860aa3457e814967a2f304fb07b446c1 : SUCCESS in 8m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24fc8d1f61f041349f8f874949796da1 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e75125542e124a1eb005ed35966db5a7 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad1881a45a714757aa3e5a44dc59f9de : FAILURE in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e2deadca8e1488cb133f05ea9dd83e8 : SUCCESS in 9m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/68ad9d54e0464aee9c81bbca8db03893 : SUCCESS in 59m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1992db153794c9b887fd0eddb628c21 : SUCCESS in 59m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/487c84b1317d4c458add5e25359b11fc : SUCCESS in 45m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a60a36ecde1f40da9602fe7457e8d739 : SUCCESS in 43m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/113ad8013bcc4c1dad5c19977bd9bbc2 : SUCCESS in 53m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5e60acbd83ce4732aba33e758c980e0b : SUCCESS in 28m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/36137bf1333e4b12b73fccfd5b0b8345 : SUCCESS in 27m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/157acebbcbb348ca8421c02da0dda557 : TIMED_OUT in 2h 05m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0752382fa90360bae1b62886e7832e389ddb1e7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 15:27:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c9c74199fb30a2f01a27094d40d2a34d795bc7af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 17:42:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd07bb174acc475293a4b99e1fc6cb12 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f072b8806b841afabf9897c9484455d : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09dae3fcb3664d8ebb9bb16bbd061f54 : SUCCESS in 9m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd037eb37794ff0b152e5b0676a318e : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de06c2533bd4a14acdb1d452ee48714 : SUCCESS in 7m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f4984e055a34b96b3bf8742fb464698 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61064d9eee354d9c83f3698da322f7aa : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f0c1106eb7d42bdbc8c1cdf02836d4d : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7836c4c60fef4304858c00deec5f7471 : SUCCESS in 9m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/49289cc6dfd2465a9d913262c2dd6f2b : SUCCESS in 52m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8909d2cf25a0476faa3321f30c698e67 : SUCCESS in 52m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86d3708819274f709f3d1f1fc08735ef : SUCCESS in 46m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/589cd0b1aa2a4158b1c04d67c2640e6b : SUCCESS in 1h 03m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/694dfa2dae4c4bfa99ad3a41411df7cf : SUCCESS in 57m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/875d48c0e54243b8b278389fb62026ff : FAILURE in 31m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/21c329b9501a4502b37501ecc63d2c27 : SUCCESS in 27m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/399681b63a9140c69422f1616f1f7eec : TIMED_OUT in 2h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"222b05ed9c9a364d20760aa2c8a3701b858aa60b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-14 19:10:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c05ae75dc76d48c23dce11e7f37b823b0d2c8638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 21:18:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfb5a9487344fb899d151c80809a965 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49a27db38e4247818dc30f998072c9d0 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b671d7df284e76bd0fc1cf1592003f : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/291ada04d03a498bbc63f34aa5fbd72f : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d8b523af6b436abce4bdad793d6788 : SUCCESS in 7m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea11e879992d4295bfdd1a4b70f392b8 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0667d30420a94e559a517cd71c011201 : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/683275cb76c14003bbe503ae741927e9 : SUCCESS in 16m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/139327a779254471a765ba45baa68adb : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f14a8fe6db8a47d3a5a3c10dabd550ee : SUCCESS in 2h 05m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1f88590b194b11978c30d3a7916805 : SUCCESS in 53m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3677c11d9be5446999789b046dcd619a : SUCCESS in 42m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a83ed1184f4641f6bba86ffec5495cb9 : SUCCESS in 55m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ada2b134c0c14138904eeb102262ff1c : SUCCESS in 52m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3f50448ed59e40c181959d113e8b1054 : FAILURE in 43m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/deb39096df204595a5d9066e176567df : SUCCESS in 26m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a1e69cd5b8bb4ea6815a55bbd21e0cf1 : SUCCESS in 1h 56m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38cd57549fc51a25ffd89f18ac043327f61fa456","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-15 00:57:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9fc61fa35089f2e9299186567d7489c01587e5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 02:52:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543933b406f1427ca0d2f03f577114f8 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81bc255883a042ac930b710f7ca45629 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/096f8554d5a9454fa95947cfa8308c66 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95c13925a77c4445af718bafc40d7c9d : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408008c807f54056ae3daf285701095c : SUCCESS in 6m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c882aeb155042e4b54efc4fa2027158 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ccd1798c53a498fbbc5377ec2a647e4 : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7959e80579e74847ac65d0d2571dc3e6 : SUCCESS in 9m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3930b1701a344843be47c5ee7ea2f32a : SUCCESS in 9m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e89ca4fbb1df4e5d91286c9cd4b1f177 : SUCCESS in 35m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e3710023cd7453385614d398c227fb5 : SUCCESS in 43m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6f4322101b544dcae6575f9d35901a7 : SUCCESS in 41m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/38ea11bfb31a4512a64200059dc3c9d6 : SUCCESS in 39m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7a1fbfefef5843aca1c100bb4ff910ca : SUCCESS in 40m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/10110780d30445c58e04f6c0f3c3da47 : SUCCESS in 28m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f058cc988f2c493298c4b413ce43e432 : SUCCESS in 26m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c7a9b8c8191f4402866fc1a7848e4581 : SUCCESS in 1h 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"132be3a7d472ba6de79e7278dc5e03752465497f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-16 03:47:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a526f51e8bb9489f5a05c2cfe26ac853b72780a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 03:52:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de4bee8001c9c74e2903e3bff830380e541bd333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 04:52:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6e03159a14429b8eb09f76b5249745 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c0abbee1904efbb21371f9a6d265a1 : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b575ffef1d546a09f81ec80483950b0 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f3df14088d4c7fba09fe1cb3586aeb : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2c8ea3de813469cb8a693c8885ce4c3 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7676b6c9f0c640669cc3c58cf34d705e : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e79f81ac64f4404a5e043280b9a9895 : SUCCESS in 10m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4654efe660254b688a7da5c92408d36c : SUCCESS in 46m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f50d670ce0c463dae32819421dd639c : SUCCESS in 46m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc9553e7996476a879da4e58dc67356 : SUCCESS in 47m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/309955cc8aea400ba7133bb5d33510e3 : SUCCESS in 56m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d96f21180f5a4fe68689a8c5ed728f77 : SUCCESS in 46m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8044e03eff314ba4b52794e6b8ef8830 : SUCCESS in 28m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8a0d5704b39b413eae10aa851a3b5474 : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6ff9ff4578315f6ef422e40b480bd4cba44628bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 04:52:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea7868cf43da3a7c2dc85f1c1909b502d8c9a718","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 04:53:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82058b406299463fb5a4923a9ace99f8 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c5f8527f86346b891eddc572aeda158 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5","revisions":{"d3897823f1c5a1d738075ef2d344d263da17cac0":{"kind":"REWORK","_number":1,"created":"2020-12-14 09:24:38.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/06/766906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/766906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/766906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/766906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/766906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/766906/1"}}},"commit":{"parents":[{"commit":"bce0747663df1e1d7fdc05147fe6361cccf6ce16","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bce0747663df1e1d7fdc05147fe6361cccf6ce16"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-14 09:09:58.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-14 09:09:58.000000000","tz":480},"subject":"Update doc8 version","message":"Update doc8 version\n\ndoc8 only supports python version less than 3.4\n\n[1] https://pypi.org/project/doc8/0.8.0/\n\nChange-Id: I065a83788bb051c2505ae6ad985b649ecc191f23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d3897823f1c5a1d738075ef2d344d263da17cac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d3897823f1c5a1d738075ef2d344d263da17cac0"}]},"branch":"refs/heads/master"},"b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-14 15:27:09.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/06/766906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/766906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/766906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/766906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/766906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/766906/2"}}},"commit":{"parents":[{"commit":"6b0ff4b6c0407bd6376398209e2e315645d5a132","subject":"Delete redundant print()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6b0ff4b6c0407bd6376398209e2e315645d5a132"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-14 09:09:58.000000000","tz":480},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-14 15:27:09.000000000","tz":0},"subject":"Update doc8 version","message":"Update doc8 version\n\ndoc8 only supports python version less than 3.4\n\n[1] https://pypi.org/project/doc8/0.8.0/\n\nChange-Id: I065a83788bb051c2505ae6ad985b649ecc191f23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
