)]}'
{"id":"openstack%2Fmistral~768023","triplet_id":"openstack%2Fmistral~master~I3b9aeb3379a76f7e40dab0c46e27f4447a0c3d03","project":"openstack/mistral","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I3b9aeb3379a76f7e40dab0c46e27f4447a0c3d03","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2020-12-20 23:54:07.000000000","updated":"2020-12-23 22:43:07.000000000","submitted":"2020-12-23 22:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"768023-policy-json-to-yaml","meta_rev_id":"255fd10b7c4140e0aed79f0a6f459c32423c1311","_number":768023,"virtual_id_number":768023,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 22:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-12-23 08:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-12-23 08:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 02:03:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-22 09:01:49.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"CC"},{"updated":"2020-12-23 08:19:39.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"e23c403810ac351f4a7b3910d16e426f2a8ae9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-20 23:54:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2de6cbbe1c6795eaad8985b51900acd6d9c4bfb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 02:03:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8937f653b1594ceca718150edbcbd975 : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5ba84c13c44a089828e104c949b271 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26d3845ad4ec4eba9d5b770ae41eae9a : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68b6a72d0a594132ac3cb0dd5b2bb2e6 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b219eb8100494b269f54a6e29f10a429 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2185d6d98daf47dc97ead8b74fec2ffd : SUCCESS in 7m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3fe1edac3e6e47f09d60ca95dc462f4c : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01bdce6d19e548d89fffdbee34df15ee : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/151fd091656345e39985c245c9603427 : FAILURE in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f56d198a05411ba93534a4bb8f66ca : SUCCESS in 9m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a84771c5fa474a09a654b9bfe6e25498 : SUCCESS in 45m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2ea7d9bd567474ca95ed97b4cdb5424 : SUCCESS in 37m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62d67440691444fb8dd1f3f1b0f62cee : SUCCESS in 34m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8a5a0df03c06415f9dd8a3043a1e5015 : SUCCESS in 32m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/27804ee9597744b4ab089bfb925758b2 : SUCCESS in 30m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/29184dae85bd4260b09d8f1344c01e04 : SUCCESS in 27m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b8c3fbd2771e48c88a1ff7a755b5ed2f : SUCCESS in 27m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/cce36ae67b714c80835649609875eee2 : TIMED_OUT in 2h 07m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55f14791d38768f5a04adc0f352c8208da48e3fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-22 08:33:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c6c82a5f370f5ef7a6443ed53f414826966c12f2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-22 09:01:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1bcaa0e9dcad6bd65c62ff18a212f66959f0ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-22 09:48:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"866bd193c87bac3f699f80470b7e378e49ef5826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:56:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6803ad3dfdf5440580474886a9afe36c : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1531df8fa624004b82c05654d6f5613 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bdd3d7704d046f29969eb7a2819afd5 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b21129ca69c41b9b931f105ec4c6202 : SUCCESS in 9m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14b4beb51aa7445ba22c2f5cd5ae3b89 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524d0940c08946b29ab340eb02bb3694 : SUCCESS in 7m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/890ec9fb47e245f9ba95a32130ee5773 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74523a81d2824165a397e46cf4bcfe7c : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e5c009b7b1b451eb8812ccc4c5e108b : SUCCESS in 8m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f40becbe433d4305a2eb59309e1ed147 : FAILURE in 8m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4a40934af94b451a99553f18d7ecc0a4 : POST_FAILURE in 11m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/296a89ffe83041b4abe172a7260e22fd : SUCCESS in 47m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfadc6daa08445c480c37a0a7e696d53 : SUCCESS in 42m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bc9f7813480f4f6c852f57e1d7895215 : SUCCESS in 42m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fa1ba84e0abe4cd5a15d25b39d33b278 : SUCCESS in 34m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8e74cb74137f45ffa6e879f956f483a3 : SUCCESS in 28m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4a977f9201cd445d860649f611e4c6d3 : SUCCESS in 26m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0dad3846109d4053a561a1ed548b7bee : TIMED_OUT in 2h 04m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d4a2914941fdc8436adedd0f92057d3f010b855","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-22 14:05:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac18b23c5c155421b319702cbc64c6d42ac2b4b4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-22 14:05:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4e0f5fa0a16310ecbd90cbfe56172dc850daff4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 16:15:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1797bd1160de4fb48164b921f0b70f01 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6795c3e0bbdb449b9cff11e2ce74811f : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/450987be86674e2ea570972e87d31795 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/067902e94213484395568f2a77678786 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b652281f6c04cb2824d8d3e7a0879df : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc809071fa8419380a2263e65347dc9 : SUCCESS in 6m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0cc85d3a21e43f592eb92d010bdb23c : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef499151c2b247de85136789e58e923b : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff3e98fc6936490390ce4b1cb8d30938 : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a968cc75904206af484f6f3cd6b3a4 : SUCCESS in 8m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5ea86a4c11d649d0a18227625f582fb6 : SUCCESS in 49m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fe5b81b7d1a40a5b4850ed0759644fd : SUCCESS in 44m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d43a88309bb418881f6e79c124e441e : SUCCESS in 48m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/40fad9fdfc97488cbf40cf6aee49d946 : SUCCESS in 54m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/37179a11f64c4e45a54712dab4da80b8 : SUCCESS in 48m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d83e31a94fe44ca1aa73fea8272d5d0c : SUCCESS in 29m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e73855c2698e4f3bad8d320ea8a74e45 : SUCCESS in 27m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/10f7bcc0e3114564aff175f8147a37cb : TIMED_OUT in 2h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cfd8a4e91b7b2ae1003ce6dc5edb572e7aaa1de2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-23 08:19:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e6015deac6f74b802659604e0b0d4a277b277a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 08:19:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8168f32bceae2bb07286e5722a198bc0f25f6b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 09:03:30.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bca74ec400a14eaf91838ec250ef1b10 : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed7efe99c31482d84b6a894744c3d37 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d292a7a3ae234ce7b4d94777d48b7d85 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4acdef92230b4ebdb2159e2f7b631fb1 : FAILURE in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f506408fae12451bb23976495d564712 : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15a10b25fb6c4646923cd2b63058c1a1 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6fd8ab6f91b474b844060de1fd57379 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f799a9b730b844f390c48a7d566a89d7 : SUCCESS in 10m 00s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/35d0cc7a2e024042a34e9fa1b50ed325 : SUCCESS in 39m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5aa51d3ca2340b08e085827d148f870 : SUCCESS in 36m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a19b81b5fce240b3a4797d7907764541 : SUCCESS in 40m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/49be3996c7d1450a85fdb3f622bcc352 : SUCCESS in 39m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/89196b3884194dcd9de1d625b8218710 : SUCCESS in 42m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3f166b5b330b4aa6b9a23032b901c7ce : FAILURE in 28m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/502d43a7674147238154b65cf301de6b : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7aad55aed6b88ea2ec4afe56191f0b3f0e8fcf24","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-23 19:51:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"70f7d18b11130af5009f8201d72a40afb87bed37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 21:46:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/09ba3381d12f484590ef3b91dbd282d8 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dad9a9af20047f086679eb21e5c3b09 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa5f9ceb1b684656837a56f3bfbc3575 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c159ecf5817145cb959256023ad9ffc7 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d29d12cf7a74d4ebeb53014e001f13c : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fa2625200f4652b157cbc25ea03fae : SUCCESS in 7m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0c17ce4c9d94d72a187164bad68f03d : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae186f01a0084adab18170f6f60bf362 : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aae68de097c4fc0a716852ac6e089c6 : SUCCESS in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ead8bb374b694fe9a3317309cb64935d : SUCCESS in 8m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fd7a0524d7f440ae9c3b105fe8870d6b : SUCCESS in 41m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59a29fdd71b9421c811ace98ed37f0e8 : SUCCESS in 40m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efc7bed236b54d19aead83bb3d40d442 : SUCCESS in 44m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c55a88c85514ecd8ab0a7a70afd9122 : SUCCESS in 48m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/baa2a3fc1e4e4fd1a559a9848eaf7ffc : SUCCESS in 42m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5351780ab89b4872b6826fa059721f6d : SUCCESS in 28m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/72d80da2c8924ecd84843ae532538da1 : SUCCESS in 25m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4a6247570bfd43ef8d4aaeec16762689 : SUCCESS in 1h 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34dc7d8573ed7af4fc591d18ab3d77cfd8836c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 21:46:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"67766fd21520cfff6ff5604309df47990b54a07c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:41:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b97a7aabe2734186988577787d4136bb : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d97b758972e41c99797dacf8cb74145 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b30ae06c33d64b6d8bad1b61843b0079 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd6effba29184c10a0f7b73a2fc76d6e : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b41e34c19ed4f68bb9af6507a3db3de : SUCCESS in 6m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef5b54014c644da4ad0c7f853d16570f : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/693d3d96aa474707b5b2cbe9b1afc0f4 : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c9679df80ad482881d3ad7f9b5782ba : SUCCESS in 7m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ef3663b043ba4c48add3f37720b14824 : SUCCESS in 39m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37c9e8f93f414f5fa9bb4eceadce89d5 : SUCCESS in 53m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9347f13ecb9247a4a00e785994c42de5 : SUCCESS in 31m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d3154ec7f3a44556910feafe5ec9c359 : SUCCESS in 31m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d9dafead48574e20b32262e53afcee52 : SUCCESS in 40m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9a22391a188a4c948519a96cd10ea910 : SUCCESS in 27m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2d20b448d95144ab83b8fe4944c06064 : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":3},{"id":"897b8a3526b9a8514874c7041750b9cd16dca5fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:41:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"255fd10b7c4140e0aed79f0a6f459c32423c1311","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:43:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a7798f2fde4db49e288a2c92731743 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a78062b6df374a53b67db895acbd6742 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d","revisions":{"fd257285c7409410995751429b42ba998187b0c6":{"kind":"REWORK","_number":1,"created":"2020-12-20 23:54:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/23/768023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/768023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/768023/1"}}},"commit":{"parents":[{"commit":"b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5","subject":"Update doc8 version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-20 23:53:14.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-20 23:54:06.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I3b9aeb3379a76f7e40dab0c46e27f4447a0c3d03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd257285c7409410995751429b42ba998187b0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd257285c7409410995751429b42ba998187b0c6"}]},"branch":"refs/heads/master"},"45e59cde998202d8f25f1e13c1c9854c14ce3f6d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-22 08:33:58.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/23/768023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/768023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/768023/2"}}},"commit":{"parents":[{"commit":"ef162b9efabcefe7cceda8efbc2c5380bab7bd25","subject":"Fix version of requests in constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ef162b9efabcefe7cceda8efbc2c5380bab7bd25"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-20 23:53:14.000000000","tz":-360},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-22 08:33:58.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I3b9aeb3379a76f7e40dab0c46e27f4447a0c3d03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/45e59cde998202d8f25f1e13c1c9854c14ce3f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/45e59cde998202d8f25f1e13c1c9854c14ce3f6d"}]},"branch":"refs/heads/master","description":"Rebase"},"7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d":{"kind":"REWORK","_number":3,"created":"2020-12-22 09:48:11.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/23/768023/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/768023/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/768023/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/768023/3"}}},"commit":{"parents":[{"commit":"ef162b9efabcefe7cceda8efbc2c5380bab7bd25","subject":"Fix version of requests in constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ef162b9efabcefe7cceda8efbc2c5380bab7bd25"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-20 23:53:14.000000000","tz":-360},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-22 09:47:46.000000000","tz":420},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I3b9aeb3379a76f7e40dab0c46e27f4447a0c3d03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
