)]}'
{"id":"openstack%2Fmistral~768153","triplet_id":"openstack%2Fmistral~master~I5696e74e58ff3594a5c65e51c12652c2a3e62038","project":"openstack/mistral","branch":"master","topic":"fix_low_constraints","hashtags":[],"change_id":"I5696e74e58ff3594a5c65e51c12652c2a3e62038","subject":"Fix version of requests in constraints","status":"MERGED","created":"2020-12-22 08:13:37.000000000","updated":"2020-12-22 11:09:43.000000000","submitted":"2020-12-22 11:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768153-fix_low_constraints","meta_rev_id":"f492d862d45b6b93f1f1d119a74ac6d5b86fc2e9","_number":768153,"virtual_id_number":768153,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-22 11:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-22 08:29:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-12-22 09:02:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 08:28:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-12-22 10:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20636006d9befe5076e41f27e2a02055ee4bdb5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-22 08:13:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b190ae0afc7e94d466268e3eb33e46009ecece0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-12-22 08:29:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"941a1eaba8218110aec932449fd4c18cdab7457e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-12-22 09:02:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9af7ba4bdf626a07ad57261ba971b09f53c6647e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:13:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4ae9ad695db4dc89db7ab1370687d93 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a674ca130c04efa918a8272114a7165 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6c0c92670004e09a8b0511de3fc96ab : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b64516069fa140c1ae2e1eeffc3e6f39 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6446402799a04dbdb8e951ba020ecd2b : SUCCESS in 7m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/699d5dec3b8b42069e960066b0465cbe : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57c35cc58bf846378b27d61c07a4c776 : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff49910bdbb484a879259790c726288 : SUCCESS in 7m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/861055764713477691c9e3f2e441f872 : SUCCESS in 42m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d36035758265464e9483b245a646e7f2 : SUCCESS in 42m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77af7c3347bc42aeaaeef2dcacb879a2 : SUCCESS in 42m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/06aa0c7e5f84410facbcc553afee7f5c : SUCCESS in 42m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e56485dda3b24c41a06d7d69d1a51b35 : SUCCESS in 42m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1abc4d77500647348cf19049516a1e81 : SUCCESS in 28m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d5a8bbbf611943a185105bfbef6b1212 : SUCCESS in 26m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b6fade273cad47c0ab555e81649366d7 : SUCCESS in 1h 58m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b84a2dd5538187ee0b98930ab9b4e9eff09e4f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:14:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"faa720b1f5687d25d0109ae4992b52702e6d9b39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f89d76c7a304830bdf9c83cada2d1b2 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/322724bcc3504f45b3499cef60977f2f : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33e2d57b52604b918340366f2f4ec845 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ae5252046742a59fae9dd3b1350dc0 : SUCCESS in 6m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7db71da8fae74b91a2ec81389b3cc33b : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9bc9ef5e6a74e788e97d729046678a9 : SUCCESS in 9m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c312037c1e444b038faa954bdba84ebb : SUCCESS in 44m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b584c9761b942ecb9496c0d04aeac62 : SUCCESS in 41m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7968775a921f454db497154f2afbb633 : SUCCESS in 45m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8fb062b2156a4a7490ddcf07df4f60d0 : SUCCESS in 42m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/72c4fdc63e354078a7fc88ae49d02322 : SUCCESS in 48m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8e305095bf1e40abb5f2d2011b3c75f6 : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1a2b636cc0f24d6eb4a891df07e964ae : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f310be24e4caca2cde272e69ada4123b7e005f2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:08:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f492d862d45b6b93f1f1d119a74ac6d5b86fc2e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:09:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67cec6dd1d984467850e786331d5e834 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef162b9efabcefe7cceda8efbc2c5380bab7bd25","revisions":{"ef162b9efabcefe7cceda8efbc2c5380bab7bd25":{"kind":"REWORK","_number":1,"created":"2020-12-22 08:13:37.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/53/768153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/768153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/768153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/768153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/768153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/768153/1"}}},"commit":{"parents":[{"commit":"b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5","subject":"Update doc8 version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f609d165aeea5aefc4ae4f4ca153846c6ed3c5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-22 08:12:52.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-12-22 08:12:52.000000000","tz":420},"subject":"Fix version of requests in constraints","message":"Fix version of requests in constraints\n\nChange-Id: I5696e74e58ff3594a5c65e51c12652c2a3e62038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ef162b9efabcefe7cceda8efbc2c5380bab7bd25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ef162b9efabcefe7cceda8efbc2c5380bab7bd25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
