)]}'
{"id":"openstack%2Fmistral~768922","triplet_id":"openstack%2Fmistral~master~I8bea8e24aa5c12fe08d3705f042bc6dea8695403","project":"openstack/mistral","branch":"master","topic":"remove-unicode","hashtags":[],"change_id":"I8bea8e24aa5c12fe08d3705f042bc6dea8695403","subject":"remove unicode from code","status":"MERGED","created":"2021-01-03 02:54:03.000000000","updated":"2021-01-20 10:51:02.000000000","submitted":"2021-01-20 10:48:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768922-remove-unicode","meta_rev_id":"467d7f5c2f96abad60d1270ae23522ba521bb10b","_number":768922,"virtual_id_number":768922,"owner":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 10:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2021-01-20 09:34:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2021-01-20 09:34:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 05:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 09:34:06.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"d55281121b980dbd6830d03037fe88be128b7a86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-03 02:54:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe0ec64deda76e7a8e1840d2f9b55a5b1a348549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 05:02:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b076ec8f26cc4b4b854c5b2f058ca5c0 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b84e5a541824b40bd6fd45a631cfb0e : FAILURE in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e59647547974d7ca5a79e82671a7ce1 : FAILURE in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc7783946ccd4ac5b9118a654f7ba4d9 : FAILURE in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72aa0ddd89d0489e9c5a80d1ab5502fd : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5300576b35140349d2e49085e01b51a : SUCCESS in 8m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9f5a27fcfd940d9b8d33db340ac5272 : FAILURE in 7m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/85bc6bdf16ec45da99f182ec3dd32dcc : SUCCESS in 43m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17feb0dc5df548848fc591e6d9771cd8 : SUCCESS in 44m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a073b17f7dad41769767131c64852da2 : SUCCESS in 48m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/abcd7abc6e864225b8f67673fbaca861 : SUCCESS in 44m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6b8c2d23e8fd4debb843d1886d80cce8 : SUCCESS in 46m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/42ce0892373a4e948936dab95d50f064 : FAILURE in 27m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f314e97835114f3fa97293c992c51470 : FAILURE in 26m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dd9cc215139447cf9a57a43c6d45e069 : TIMED_OUT in 2h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82b034c635d0c74f7995e992030099f364760a8a","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-03 07:39:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"86e6475e71b6000575cf11b18514f9092516a7c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 09:45:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/165f62c297ca4f6cbcc21066619906b6 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a99654fbd1b14739ba82a08d06d2de88 : FAILURE in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fefc31ce79c44f72bb502efdde8c61be : FAILURE in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1de7d5de56d24be48c17962c0967b586 : FAILURE in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3d21b4fbc4429a8cb139d609d32077 : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e1178b9986546e78489a8d63ff4e3a9 : SUCCESS in 16m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20b998a964a145da8c52e5afa614bf1e : FAILURE in 8m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/87de068823db4de09b5c13fc3c11c00d : SUCCESS in 39m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d423bf64e733403eaef6575b4fb898e7 : SUCCESS in 46m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13e7abc4ac8a40b1bfb0a3adae8b5bb5 : SUCCESS in 55m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/185fbcb2a31447e0a93ca62d659434d6 : SUCCESS in 44m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/73d6d56b0cf14bf7957dbc9b705d1120 : SUCCESS in 50m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c83e49da038e46daad1fc7cd51fd60b9 : FAILURE in 28m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b5db97a7c34944dbb4e01d1638ab6f39 : FAILURE in 27m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/030b31d900324f32a90b53cef4c9b12a : TIMED_OUT in 2h 03m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acb302001a7c52a5562753aadbb8e08b9b1543d5","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-03 12:10:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94e8a480419572178c47b00588d74d2ad56ea28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 14:10:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/104d29dff412458a8177eec4d1e59eeb : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7830964ae60f408d93b8d0dcbee7787b : FAILURE in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b19e97fb6a334cc8a1bc6a0ef17b2fb7 : FAILURE in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b5d9cf4efc64ee8b6ada29c097d512c : FAILURE in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785438ef2d994624b1c0f0c60821a4ca : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5e961db7aa341c086c328c95538b3c2 : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffdcb0f4581b4d6b879cf54fb17f8cdb : FAILURE in 8m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ce804962fc00478083f1509ac60fd866 : SUCCESS in 47m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c956060324d445589a0f815c83917f2f : SUCCESS in 43m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ecf08558785405daea599c6e67ead36 : SUCCESS in 50m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fc1756bd0e304953b357be1e32a39caa : SUCCESS in 43m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a22eae3836474772807febaeed6a1b25 : SUCCESS in 50m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6d75737008ce4100861ef98fd2b08339 : FAILURE in 28m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/55ffcdde13a44a5399410a01aec51a75 : FAILURE in 26m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8aedafbd4f5642ca895c294f3977d762 : SUCCESS in 1h 57m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a75bc59a12939330233f3ffb8936185b8505030c","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-05 07:47:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ca5846bb22374b96a28c0089eca06878ab6d9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 09:56:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcac4f7e3e214443810a4351cf48187b : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c03b0d297e349dfb3c382eb7732b28a : FAILURE in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e97ccd373036447ea0d0208b641f078d : FAILURE in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8e3331b59b8404c8dea24b52a0883dc : FAILURE in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e197c41c3c4b03962af5552224b9fc : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c73a2f61535c47228d953d1860a7b3b6 : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cde8a4c893d4988ba8525a2b69e6d9b : FAILURE in 8m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/63eeaf9189ad4fe4be8f7cc973f07d49 : SUCCESS in 44m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c17cf43d37a24a468ed06499af323172 : SUCCESS in 48m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/418299464cb74a68b07164d5d7c389f9 : SUCCESS in 44m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c174b99620748988d72ddfb5da64ed0 : SUCCESS in 1h 06m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/16f8586e7f9142beb476213e3bd710d5 : SUCCESS in 49m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c5fd1a3bc29c443dbc37904388ee0b3b : FAILURE in 28m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/885588c793d443ddbe9b6e9454fb2d65 : FAILURE in 26m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e226758a483d41648320060dac6ab323 : TIMED_OUT in 2h 04m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e42c1fe51ad9c80951e7b43a0da6bb678b2cbb8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-12 07:26:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49fe013fcb0423816c05be41d6a52941efcebe2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:12:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647fd1b1284c4354b827b7c13c489190 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acb602a4abe64480b1e3efe4b6260a0d : FAILURE in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31ce570638d54940aada055e2ae45a6a : FAILURE in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b332054a1784f4981b44c54dbe3e11f : FAILURE in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f847a617814bf9a9840fbfbf3913f3 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55b585cc52974b66b6d4040e2d567d9b : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/301e0137124c4473a36defb0e770b592 : FAILURE in 8m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/29ba051140f84a81ae3414d7dea20b5d : SUCCESS in 45m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9fec349173c473ebe33e542156d224e : POST_FAILURE in 54m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4ba22903eec476297640a71b042ac41 : SUCCESS in 46m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f552e162317a41edaf6050fa894e830d : SUCCESS in 46m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d4a3999a8f1b4372af3b244c3614f37d : SUCCESS in 50m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/09bc8162930b4fae943dfea5ee8dff5f : FAILURE in 30m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/685efcfad1c842f59de39b52a2fb5f7f : FAILURE in 28m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/753035aa8be247c2a4b32a4db0c576bf : FAILURE in 1h 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f97b2c96d486185847d073e9326bfb7617686d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-01-20 05:10:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"467fab7c7b9a189ae480f8c9b424ee450ab4272e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:31:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffde22c3e93244ed8074e576c0dcae6c : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9838ac2f6e924766998d9955f64d49b4 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d56698ac74874906ac432b09b98f49fe : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b98bf2e016e840feafd19222a93a0bc1 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e5103e0c354c769aa3399a9273b1d8 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b81d502865f433e964de521952e5e62 : SUCCESS in 9m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44cd60069be8404489e46354183bd4e1 : SUCCESS in 8m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b45e1b23d7634fc4aa8fc95a6d8f70fe : SUCCESS in 44m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbeaec78ef3e466dadd824b953990061 : SUCCESS in 50m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c75f05d7ffd24e1f94cb4d2f038ecfaa : SUCCESS in 44m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9b4ed5df84c74df3a1ff07b8ff5bc364 : SUCCESS in 50m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9c69ba64c79e4c429be291f6e3d01d62 : SUCCESS in 43m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1d22caebd9c044b99b6083f1d733fa5c : SUCCESS in 29m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/782b225882984515befa8ae77df15baf : SUCCESS in 26m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b6764f7f28034b1e90c2d491fa16b0a6 : TIMED_OUT in 2h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f828fdf3f9900c2b810dae00b20bc4ff38e7ca06","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-01-20 09:34:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4f72d21c6edcd882c4f91a85c825dc33aa2c59f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 09:37:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"05356e0a47162f5bc8de5ef002c1e4eeb335a9f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 10:48:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3331178dea440f8b72a43e494043a90 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23551dc0548f4fa3a0b56606c1287c6f : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/278b0c171cf2453aa3c4958bd042d02b : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f80f36fd8cf470fbd4b5c7ea40946d2 : SUCCESS in 8m 46s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b065b7af877f4f039a4331831c010eea : SUCCESS in 1h 06m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98cd26cbecc740bcbcb96414f6197af6 : SUCCESS in 48m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cef28490056b46fda955400f37d07aca : SUCCESS in 45m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5dab35ef362f4500aabf8650a7ab6f24 : SUCCESS in 50m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bcf0061bc079423da4e933362a806d9a : SUCCESS in 47m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/44309cdaadd1411daa4591a9ca2dd75b : SUCCESS in 28m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df5b1ba15b8a4aafa95f47f9491f6f26 : SUCCESS in 26m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6b4b1301134f2e6350b345cc58b97c6d9e2d7f73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 10:48:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"467d7f5c2f96abad60d1270ae23522ba521bb10b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 10:51:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdafc5553b2a4bd08339dfcd314e5153 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"de4d0277e0ef70330b491caccf70c833f92cbd57","revisions":{"8ab3e42f57a55f8faac257205804e479d7164ad8":{"kind":"REWORK","_number":1,"created":"2021-01-03 02:54:03.000000000","uploader":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"ref":"refs/changes/22/768922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/768922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/768922/1"}}},"commit":{"parents":[{"commit":"7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d","subject":"[goal] Deprecate the JSON formatted policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 02:52:17.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 02:52:17.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I8bea8e24aa5c12fe08d3705f042bc6dea8695403\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8ab3e42f57a55f8faac257205804e479d7164ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8ab3e42f57a55f8faac257205804e479d7164ad8"}]},"branch":"refs/heads/master"},"ab6dabed87841587e0fdd3b8392aaccb81cb728d":{"kind":"REWORK","_number":2,"created":"2021-01-12 07:26:33.000000000","uploader":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"ref":"refs/changes/22/768922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/768922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/768922/2"}}},"commit":{"parents":[{"commit":"4319dbcb361a6b63ba241d4be0f7d684e88455a1","subject":"Add API docs for code sources and dynamic actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4319dbcb361a6b63ba241d4be0f7d684e88455a1"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 02:52:17.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-12 06:28:58.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I8bea8e24aa5c12fe08d3705f042bc6dea8695403\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab6dabed87841587e0fdd3b8392aaccb81cb728d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab6dabed87841587e0fdd3b8392aaccb81cb728d"}]},"branch":"refs/heads/master"},"de4d0277e0ef70330b491caccf70c833f92cbd57":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-20 05:10:23.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/768922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/768922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/768922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/768922/3"}}},"commit":{"parents":[{"commit":"421f84334b685e3c0ab6b06c8b904db891291993","subject":"Fix a field name in the CodeSource class sample","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/421f84334b685e3c0ab6b06c8b904db891291993"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 02:52:17.000000000","tz":480},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2021-01-20 05:10:23.000000000","tz":0},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I8bea8e24aa5c12fe08d3705f042bc6dea8695403\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de4d0277e0ef70330b491caccf70c833f92cbd57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de4d0277e0ef70330b491caccf70c833f92cbd57"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
