)]}'
{"id":"openstack%2Fmistral~768934","triplet_id":"openstack%2Fmistral~master~I66616791d05e278127c7daa8d0644ea0d2c11fb6","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I66616791d05e278127c7daa8d0644ea0d2c11fb6","subject":"Add py38 package metadata","status":"ABANDONED","created":"2021-01-03 03:08:49.000000000","updated":"2026-03-28 19:40:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1eba954a706f455f6a26a472ae682a640aa827c1","_number":768934,"virtual_id_number":768934,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-06 11:22:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2021-01-25 07:50:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-06 07:34:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 05:16:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 07:50:36.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2021-05-06 07:34:56.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"ce2deeff7d7a8a505bd83d1440fda39884fa316d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-01-03 03:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7308c48b932a62737b92b39aec25cc96afb1abc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 05:16:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54c245fb05641579a0430ff2bde7b01 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b61abaa1bd9403c92df1ab288c4a602 : FAILURE in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/493e9979eff144ba9bdd86d5d63f17db : FAILURE in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16edff25063c4e54b16e288d880f8ee7 : FAILURE in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597e61fc5c614c67bcaaf50b54fdd892 : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f47dcfe096a413ab4cffe7feee424a0 : SUCCESS in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d42c09273c654bc4839d48f366ede56d : FAILURE in 8m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3c0a9baade7f49a8b7c79a7345e6e8e6 : SUCCESS in 1h 10m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8fddd0383d2433d935318d278bf64a1 : SUCCESS in 1h 07m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61f475744cfd425da7698d5c3a3ea582 : SUCCESS in 1h 09m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6a272308dd68439b9cf2f58dbfc563ac : SUCCESS in 1h 09m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4460bf5c0e594ba58922f57a750640f8 : SUCCESS in 1h 06m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/213f578b59714cb9b112be7ba879b983 : FAILURE in 33m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/92effd3c72364cd3b418f5ab82aea3f3 : FAILURE in 29m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fc4d2cc510954f949c26b07f7d43fd30 : TIMED_OUT in 2h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fcda8945bc31080f99ce232ba4b4c7285d571ab","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-01-04 07:18:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9648fe4d635ca94febbb092e6cde319961bd9424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 09:27:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f49a0744f84d9dae79bd55599d9b9a : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baaa4eeadf3c4e91b112b5bf96baa8ef : FAILURE in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/105e6321e41d4a5fb5413aa9dbeda07c : FAILURE in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc2c5c86b5c4b84b3a8a748146a70da : FAILURE in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14eb3934a4b143e2960bebe8ccb8b0e3 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66a7d9dc16414077b9fbdfd82ad44c0b : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/634059b8fce8494d9ac1e4ec5b6e629f : FAILURE in 8m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4619f49893054f3f96f37e9ff50a4d19 : SUCCESS in 41m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0abb6b02e9c41b9896a57d063ab9373 : SUCCESS in 42m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f56f6fa9886e40299ca1c0669ef06ef5 : SUCCESS in 41m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b81a20f6b60449b9b202099675c363b7 : SUCCESS in 42m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e134b9b6befb479eb50090a767a47298 : SUCCESS in 40m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/049e101095494f19af84a8492420e4d4 : FAILURE in 29m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a79cac3782fe4b2da49e20f6bc67246f : FAILURE in 26m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c45cf336ce2d48e2b9b2c30106677120 : TIMED_OUT in 2h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93475edee2b4416c7f442c905dcfd94fed362e73","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2021-01-25 07:50:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"86589e8ab34643e667976072d2ba597ea296b57b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2021-01-25 07:50:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a868d54c0304ff5d91ffaaffe8503a9ba68804eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 09:37:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6998503d4fe1492fa02c090b15e8d339 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bfae445dcc640fd955c9004fdc29020 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/496356328ce54b45911a21af4806e069 : SUCCESS in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb6590bffeaf41fc83ff90c2a445e418 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d7497f41e7416fb117d0b211400535 : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae8cf6e95d234111878ff3217c3b7b63 : SUCCESS in 8m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/84e77f26123e457fa261dd0761f2abb7 : RETRY_LIMIT in 16m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c16ef0f115c4070a600c21c10ed406c : SUCCESS in 45m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6aaeb5ecddef41a2a057c2781daf0f6c : SUCCESS in 37m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d152690c2b7547ae9111decd568dde20 : SUCCESS in 41m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cfde0fed502c45709231b4cf3e09cef4 : SUCCESS in 56m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7e982c5b6e9741b8a6fbe74eec838ace : SUCCESS in 28m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4111d197ff364d43baad398299e90bc9 : SUCCESS in 27m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7ac907f32f7049f29c994a2f764b11bc : FAILURE in 1h 44m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3665c88324dd58a53279a86d09b2423c23cdabda","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-05-06 07:34:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4496272f542b1c79e0314bd6303bbbcd0788e87a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-05-06 07:35:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d35e918af2d4bc0a0daeb33d84ebc3365e44e211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 09:12:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ef54f7b2db49a284701308a80392dd : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa2574845d774850affcddd1f1aa99e4 : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff4231821fe845758c0f3d93364149be : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99127e04ae174ffc8323666819943845 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe79dfabd38a4beaad3df711456bd511 : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5828822108d14bf78170ab398088384b : SUCCESS in 10m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/32f2cc82c34c4a6dbd72ca09d336f8f3 : SUCCESS in 43m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98dc0395876b4acab81521ed8533ad70 : SUCCESS in 36m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49f267ba15574192bae6502911b47266 : SUCCESS in 36m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bec77a43e125403a9fd5656abba44034 : SUCCESS in 32m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5e9341523d1f4ccd8083df8d49da98d6 : SUCCESS in 45m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cf7bb77e79bc4cc78a48e8423a086ce9 : FAILURE in 29m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ab99a113a5ea4bff8e33e57d2ff65b61 : SUCCESS in 27m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f11c2bb956004b208f637221f3ccc7ed : SUCCESS in 1h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02b25f3abca60c2efa1f8e29f4a22539085e28b4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-05-06 09:15:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6adf13b94c5695915c524d43b975a03d997d45a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 11:22:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c117b40b9cb64aab9e25b312c6170773 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff552f6738ba4b7bbc81fc8da2aac5a7 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd42844519bf40c58566a4edf66a7306 : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274efe35fd8345f88181995fe765be4d : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef153ae734f447e8adc0a0de248a13c : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209eed3a82f34d75a86051147ab58abf : SUCCESS in 9m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f87f76ccf6aa4eab9f419c28f4a9b702 : SUCCESS in 37m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ffdaaa75b944cf583a0d59e598592df : SUCCESS in 42m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7999a08fbce04ddbbc92edfa6e3f194e : SUCCESS in 40m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ec55fa046c3d4bcd9df866e1fbbf0c64 : SUCCESS in 39m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3735690d42954b9e850c1fa7d6cdbd06 : SUCCESS in 39m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4fab574092fc42deb5ed97b7b2abf7eb : SUCCESS in 30m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/db7dfce806694cc1b8cd41f0b327bb3b : SUCCESS in 28m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b3c600f078c8415db5dbe93115144dd5 : TIMED_OUT in 2h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1eba954a706f455f6a26a472ae682a640aa827c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:40:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa70cb287dee9653fef086258fefa9804616b138","revisions":{"fa70cb287dee9653fef086258fefa9804616b138":{"kind":"REWORK","_number":1,"created":"2021-01-03 03:08:49.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/34/768934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/768934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/768934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/768934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/768934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/768934/1"}}},"commit":{"parents":[{"commit":"7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d","subject":"[goal] Deprecate the JSON formatted policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7cf2014c2e4b9f8e04ed23288a2afc2260e9da2d"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-01-03 03:08:04.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-01-03 03:08:43.000000000","tz":480},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Wallaby tests that include a\nvoting py38 job from victoria release, we can now add the\nPython 3.8 metadata to the package information to reflect\nthat support.\n\nChange-Id: I66616791d05e278127c7daa8d0644ea0d2c11fb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa70cb287dee9653fef086258fefa9804616b138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa70cb287dee9653fef086258fefa9804616b138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
