)]}'
{"id":"openstack%2Fmistral~779269","triplet_id":"openstack%2Fmistral~master~I3986238b0a65fb652b30a8a47448a1d02212166e","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I3986238b0a65fb652b30a8a47448a1d02212166e","subject":"Change get-pip url","status":"MERGED","created":"2021-03-08 14:06:55.000000000","updated":"2021-04-14 09:52:44.000000000","submitted":"2021-04-14 09:51:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779269","meta_rev_id":"dcd14518638dcb0853251a155e4715137d4e6e14","_number":779269,"virtual_id_number":779269,"owner":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 09:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2021-04-14 07:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2021-04-14 07:20:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 14:27:19.000000000","updated_by":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2021-03-08 16:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fe6aa3472ae80bf2e861109bcb67a99c2b36d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"date":"2021-03-08 14:06:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dba895c239c501989c7c1db5e742d5773463ad8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 16:15:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57045b3b5934015bdab2794545f7a6e : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/532069552fd745f38010b6693b61f45d : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94e5c5249a2244918f55122861298607 : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b636131fb347bd8664ffa95b94fc35 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371e0af505cb47f986cc9009b79ca79a : SUCCESS in 10m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0c89d767ba84c1b930a15990fe5665b : SUCCESS in 10m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/89f7b8854ea24d17b9d946fbe9bcb139 : SUCCESS in 56m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beb648f9266b46a59d5aa3e03ac7abb5 : SUCCESS in 44m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/212a63970252401aa22a3a4f512ddb27 : SUCCESS in 47m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/962eb8ad38dd4eb8a3bd534065ad256a : SUCCESS in 46m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8dda9955c8024aad9f420c55022b6760 : SUCCESS in 45m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/02d606a1697644448a39869b4a731e6a : SUCCESS in 30m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/63e6cc4aaa984cdc83444eada177e3da : SUCCESS in 29m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/55c6257e57ad4584a2b5100445a74274 : TIMED_OUT in 2h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af34015ec71307945b2a2a224bce6f0fbd59ec0f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-04-14 07:20:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10a2d80451a0abc90bd9c1152b642feb16825b85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 07:21:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86ddedc0fa240f7c17b888355b36b1e02a3f70d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 09:51:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ab57d05f174cd0824c09925ddd4b28 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29d2f9ff21cc4808af3a9ff0017dbb8b : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/841657ebd2074bf98edbd8ddfa57abb1 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab3c57da4114dc982b77cd99c270e0a : SUCCESS in 8m 50s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/93a83deb70a2418693b60cea1e77c5f9 : SUCCESS in 31m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68a21b720c8643d0bf772bf889a96438 : SUCCESS in 41m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b021530b3ad9467b995a6a58a0ed3282 : SUCCESS in 41m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a523cdf11e76484ab3dce75c4d898c89 : SUCCESS in 38m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4d01339fa19b4fa4932c55454cf09879 : SUCCESS in 1h 15m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a7741800bcf84936b650151f0ab7e3d2 : SUCCESS in 30m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d18416e19b36496980542531d2ab03b4 : SUCCESS in 29m 58s","accounts_in_message":[],"_revision_number":1},{"id":"cb539a1039f7e00972432276c98456fe01ca4894","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 09:51:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dcd14518638dcb0853251a155e4715137d4e6e14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 09:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c860eb995ee743a9a0b19b1993df582f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85b9b525fb8c06acc56186ea096fe383009b24c8","revisions":{"85b9b525fb8c06acc56186ea096fe383009b24c8":{"kind":"REWORK","_number":1,"created":"2021-03-08 14:06:55.000000000","uploader":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"ref":"refs/changes/69/779269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/779269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/779269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/779269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/779269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/779269/1"}}},"commit":{"parents":[{"commit":"3c83837f6ce1e4ab74fb519a63e82eaae70f9d2d","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c83837f6ce1e4ab74fb519a63e82eaae70f9d2d"}]}],"author":{"name":"Daniel Pawlik","email":"dpawlik@redhat.com","date":"2021-03-08 14:05:39.000000000","tz":60},"committer":{"name":"Daniel Pawlik","email":"dpawlik@redhat.com","date":"2021-03-08 14:06:51.000000000","tz":60},"subject":"Change get-pip url","message":"Change get-pip url\n\nThe path for get-pip.py script in version 3.2 has been changed\nwith this commit [1].\n\n[1] https://github.com/pypa/get-pip/commit/2360f025eb2a6f62d45cf99e4c6d83ba9c4d5e37\n\nChange-Id: I3986238b0a65fb652b30a8a47448a1d02212166e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/85b9b525fb8c06acc56186ea096fe383009b24c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/85b9b525fb8c06acc56186ea096fe383009b24c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
