)]}'
{"id":"openstack%2Fmistral~783453","triplet_id":"openstack%2Fmistral~stable%2Fwallaby~Ief9c13b3ee222aeb60644726cec653e65a8b3605","project":"openstack/mistral","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"Ief9c13b3ee222aeb60644726cec653e65a8b3605","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","status":"MERGED","created":"2021-03-26 18:42:20.000000000","updated":"2021-04-27 07:43:29.000000000","submitted":"2021-04-27 07:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783453-create-wallaby","meta_rev_id":"419795b04b737e614cee175048356f874956c4ce","_number":783453,"virtual_id_number":783453,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 07:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2021-04-14 07:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2021-04-14 07:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 19:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 07:15:37.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"1586a4576cebddcc5cc65d14151cb4949013132f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-26 18:42:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c3ba70560a33a78e0c72be13ca404e33404ecc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ef12abef3a4a889b8ff728f6332934 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cadb8b863df54e508cdec6da429cedb1 : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9f981d53d5741b4b3a3d2b84c00178a : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b7c4e6d7e9645c1ad8450ea7e04f2c5 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca6a56cd5624e809a5d09a9fcfe5737 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/474da6b1bfaa402080ed4b4bf15b14dd : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/005b26deabe34fb485fa64fb11d62074 : SUCCESS in 10m 18s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/759783be71f74b40b9b05f2efce193f3 : SUCCESS in 31m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/340ac1a4b3c54508ae95a6282882af60 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e12e1bd60504937c1eb2efc2ac0f2f5198086ca9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-04-14 07:15:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59f0108fca1a7c73a10c7132f9a801d842612199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 07:16:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c301d4712798170e3fdfdd9e11d45830f3ce3db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:24:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2a33cfe6b942fcac89ce4710a2ae01 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/280c18b5d68048cf80e14ecc21d429eb : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/076a46b9903e475ea8612f4a0a3b6eb0 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a915996b8524b99b442000f92fc9e19 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69c5d9509bf14bf3b4ec28ea19cc09ee : SUCCESS in 5m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/035e0c1c37854e61ab720641fe217c84 : SUCCESS in 53m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/461432fb19ca4da0a64cba3cfac9558d : SUCCESS in 1h 01m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e11d0f7d2f2b4aa79d6b983bb899f9bc : SUCCESS in 48m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8c045c620a194750b443c708f0008a83 : SUCCESS in 51m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/64282e2cb3ff47aeb6ea2044c78ad651 : SUCCESS in 44m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/49e1060ac64c4a5fb125a2f1ee2efa57 : FAILURE in 30m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e9ef712522fd42198a37cb14f902f9e8 : SUCCESS in 28m 05s","accounts_in_message":[],"_revision_number":1},{"id":"632913a3507416574deadccdfcec1db6052aa92d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-04-21 04:56:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f6c26461eee319055281292176dfbd3434341b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 07:05:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9130501d50804c32873ed10e1b0bbe60 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/613457840343429493e7a87addd28c14 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92c2b64a541a41d5a0edf324510e345a : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25c9586bdadd49698e513fccc8ce8126 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d77eeb16b74b559988ee25a4b89225 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e692771d6cec4495bce47b68b5584617 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21624f5298654fa196c684bf5936d98b : SUCCESS in 8m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a7adf992dcfe4e05a2c99ced1da91687 : SUCCESS in 56m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a30d847f93124e7f9d07290bfe262af1 : SUCCESS in 57m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bebb8b9aba9b4aa6969b8286e4fb163c : SUCCESS in 56m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/28546179fbf64a758dfc9e7bb29368a2 : FAILURE in 14m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c8147d03c239414f9fff7afe6d266f81 : SUCCESS in 59m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/edc2233997a9428382f7e1b39436dcea : SUCCESS in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ddd23d876946472d8cf3442a6268dbfe : SUCCESS in 26m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/50f44794b1884d37ab1acef6a076c5e8 : TIMED_OUT in 2h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd40e4f6f87512dfb70e4d44ba10a7fdaa7ddc3d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-04-27 04:26:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"91e13d320317b27acc14873686421b860c640387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 06:40:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f84397116e45cc956f65678cfa0943 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3606184c15f448db3c14b32ee6cda2e : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c8b5f219cd74990bd9009af0ba17465 : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3abe1cf8ea6447639392036d489c672f : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86480feaad6945c38aff06e572832e99 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c26d438b95d4b328e8ed1a31b5fe68d : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cd5b74a3e51420081de6c1ee6247e5d : SUCCESS in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/001e0d06d7ed418cb973b8e25f272749 : SUCCESS in 33m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d28876eb1e7422aae5940eea93f771f : SUCCESS in 39m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4265b0b2d85043b39bdbd954387f48b3 : SUCCESS in 46m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/78404f18ce1f4496b024aa592e5a18ad : SUCCESS in 30m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4dc9f243339346139abafb893cf8f40f : SUCCESS in 30m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a327d99fea0d478b9d08e7ee4f3f4aa2 : SUCCESS in 29m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/617d730e9d694369b13064027539db6e : SUCCESS in 26m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fd8c7db15db743ebbbeb7b7c24827a4a : TIMED_OUT in 2h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a02507fed79628360655524f7e75ae27c8472631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 06:41:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"838b200228bb64e7071278058eccba5146fa330d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:42:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511c0cb33de540eeb06ab7f81f3c4f45 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ae278a076844d249f51c24996a241e6 : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ade2875aec6e4ec6862c896321b68420 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82dc2eb7f4fa4d9cb5a05f7a24d1149e : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75ef1ab0fdf44408b81d7ff54e42a228 : SUCCESS in 5m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0f31c6cff115453a98eeb8cd92aac2e0 : SUCCESS in 40m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3af0b2231a44a77be42e423ab676e29 : SUCCESS in 37m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eafe712715144dd789e11b19530e8c5d : SUCCESS in 44m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1bcc9f3189f14d03807388ca2b586d54 : SUCCESS in 44m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b7435846de12491189bf7f1c0537ffc3 : SUCCESS in 53m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7729aa5937404afbbd58de5a10b0a2f8 : SUCCESS in 30m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/526a772c87a441dd9d33b62f9008b6a1 : SUCCESS in 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"22a463e833fc9e7205e5a6c54800894dbe76e446","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:42:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"419795b04b737e614cee175048356f874956c4ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:43:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aea5eba7f2b416ab1f55a4c93ce4f0f : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9da01150377408c86bb0d617f6ea7ef : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2a0205d56642ad920d94953079c98e17c7b3536","revisions":{"c2a0205d56642ad920d94953079c98e17c7b3536":{"kind":"REWORK","_number":1,"created":"2021-03-26 18:42:20.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/53/783453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/783453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/783453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/783453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/783453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/783453/1"}}},"commit":{"parents":[{"commit":"1c70aa02b54736c3c571667405a640cd6fb293fc","subject":"Update .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c70aa02b54736c3c571667405a640cd6fb293fc"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 18:42:19.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 18:42:19.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","message":"Update TOX_CONSTRAINTS_FILE for stable/wallaby\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/wallaby branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ief9c13b3ee222aeb60644726cec653e65a8b3605\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2a0205d56642ad920d94953079c98e17c7b3536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2a0205d56642ad920d94953079c98e17c7b3536"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
