)]}'
{"id":"openstack%2Fmistral~788864","triplet_id":"openstack%2Fmistral~master~I252f80cc3c85da0ecd872397109dd9e35406f2a3","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I252f80cc3c85da0ecd872397109dd9e35406f2a3","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-04-30 01:57:25.000000000","updated":"2021-05-04 18:01:58.000000000","submitted":"2021-05-04 17:59:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788864","meta_rev_id":"c02f6b790a2cb68b180177e89ccd0088dcaeac9f","_number":788864,"virtual_id_number":788864,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 17:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2021-05-04 17:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-04 06:09:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2021-05-04 17:14:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 04:08:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 06:09:25.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-05-04 17:14:49.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"3a30ceba782a037a133c48501fbd7da53f0baee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-04-30 01:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97ec70ba26df26b9abdca5fe5698eb5dee5308ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 04:08:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e609e2f0b4fd4431b905be443066bfd6 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d8c6d009ac4df79f1a3d4305e46ffe : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea0abed1751147e5aacc2b1447908609 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0e6fad7bd784f4b83157705f60f15ac : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6485a8c25cfd4903a8c470386d97a256 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ef2fb9aa4546a984826099d6a964e4 : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b24624b2b48549eeb6391ab787ea5936 : SUCCESS in 8m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b85b38621ac044ef8ef00e1f971a7ebd : SUCCESS in 43m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1ea138228764947935c25267e6da6d9 : SUCCESS in 42m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75bb23b16fb142a5820159a381c581ff : SUCCESS in 44m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f0a1027e1fc94902a1800266e30d110e : SUCCESS in 42m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ad9e926e667048ff8420f4ff1fe6fce9 : SUCCESS in 42m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f4f12c30e65047e1b5f03930457698dc : SUCCESS in 29m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1bee3bb5778b4a6fb22bcd00b47884a6 : SUCCESS in 27m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2996a2375fd3407caac846e6a5e04139 : TIMED_OUT in 2h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b227280a3e745d39ab1fcf45f419d8e7651f1ae9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-05-04 06:09:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea25feeadf5b8f228d2292166431b62f68e8ae55","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-05-04 17:14:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9f402dd9f4f4afd4d48ae6f78af5d25ee6fb553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 17:15:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8899e1b50ab13bb0750eb598f6cfb415dc8fb884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 17:59:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0825f5992e9a410ba4965fa4656aa2d8 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e41853099d546fa937397b76da27c83 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbc24ba1ad07408f853a34465b77bf60 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/435ae5c34fa1473681f6c34da496c8fc : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10684658c55a4759b0f7a48bf49b15f5 : SUCCESS in 8m 21s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3afceee60bfd4153997d3647ec5166cc : SUCCESS in 33m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5eaaeba1479a4bc49f1549415d0c6049 : SUCCESS in 40m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fa4f76bc6ce4d499a02a49754a8002d : SUCCESS in 42m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/06e43a07a5b347d9b8a03e32affcf3b3 : SUCCESS in 39m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5bb32582f5ea40cbb7d68e272c0476c0 : SUCCESS in 34m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bddb9c6a34364008892d136f06a3eda1 : SUCCESS in 29m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/52f46cdb648c4130936e513104faef46 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e51b54b70220a8cefe8fa0cfd2d63deac51a7817","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 17:59:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c02f6b790a2cb68b180177e89ccd0088dcaeac9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 18:01:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e7039f7edb40dd9698c91d6a06da5a : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ee7a3651028a520dc8ef2be14d765502de4d2da","revisions":{"0ee7a3651028a520dc8ef2be14d765502de4d2da":{"kind":"REWORK","_number":1,"created":"2021-04-30 01:57:25.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/64/788864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/788864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/788864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/788864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/788864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/788864/1"}}},"commit":{"parents":[{"commit":"b1ee134766db0c16c676a0952fbef4613a84fadc","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1ee134766db0c16c676a0952fbef4613a84fadc"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-04-30 01:57:08.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-04-30 01:57:12.000000000","tz":0},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I252f80cc3c85da0ecd872397109dd9e35406f2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ee7a3651028a520dc8ef2be14d765502de4d2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ee7a3651028a520dc8ef2be14d765502de4d2da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
