)]}'
{"id":"openstack%2Fmistral~793326","triplet_id":"openstack%2Fmistral~stable%2Fvictoria~I9bb15642585af22ad354dd4f8f2253a47129639e","project":"openstack/mistral","branch":"stable/victoria","topic":"action_pagination","hashtags":[],"change_id":"I9bb15642585af22ad354dd4f8f2253a47129639e","subject":"Added sorting and marker for working of pagination of mistral action.","status":"ABANDONED","created":"2021-05-27 09:14:06.000000000","updated":"2024-03-26 18:21:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"0947c570f9052de23841018bc620470ae9f4826d","_number":793326,"virtual_id_number":793326,"owner":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-27 14:49:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-05-27 10:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 18:21:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"5cbc422357507ecfd397a15c65a0296f25f3ddda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"date":"2021-05-27 09:14:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73bbd6954486084a6f8986cd7ab08cfc8c1d1c43","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"date":"2021-05-27 09:15:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"f19cd80fda00c5e66876a58f7d7cf0252435f8cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:44:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8193128e91a44df09c8fd294908584fa : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9532e8a2d8bb42ed9a6718c241cc6dcb : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de5be7732fad461f963ecc6c166fce79 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6a057f07924c13b5488c3cb54463d7 : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3a04488f7a8480d9a155b3d6894dbc8 : FAILURE in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6a6bdebae054bb482e360af0692dedc : SUCCESS in 9m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/59856484322141d3a105db7a77671df6 : FAILURE in 1h 14m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/641eb842ba4c47d38e8e713d00feafee : FAILURE in 1h 10m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9b5ede4f93e42f9beba6878b3d1356b : FAILURE in 1h 07m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7df7531dbba64c86bc7062fb8726f5a8 : FAILURE in 1h 14m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a817353da1af4b00b01f034323585853 : FAILURE in 1h 07m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/16742d7e30b842519831ff5548626693 : SUCCESS in 29m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a8e2a14c941e489f907de93f55bd77c6 : SUCCESS in 29m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a4ef73a7734f41a3a239ebdd2307928e : SUCCESS in 1h 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49f600f0ad7bf7009ecfc67ee1c61321cc5730a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"date":"2021-05-27 12:27:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb7ce0c980c6369e150bdcb5474303f61ba2624e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"date":"2021-05-27 12:39:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d707f870eef33f34f1190580a1c6018a68977442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:49:17.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a068cf791bfe49f7bec8eabb2d8cfca8 : FAILURE in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2cf7aa928744d30a2bdef89a2d0b719 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/420eb49167e740c0bb31967dd20c346e : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771693d733794acd8346cc28a0ba2bd3 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ef515fae7154a4a9b95ffeeed4911c6 : FAILURE in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a129c3e20a746d296bd06a245a1b97a : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4c0732160f344a5e9b97f1202bdc31e5 : FAILURE in 1h 12m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc1ab40dfd40490ea9df80b06cff9717 : FAILURE in 1h 09m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0b035307f5c4098ba106a1c64e76c64 : FAILURE in 1h 09m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/928bd9ea8f874912b915cb77e70c4140 : FAILURE in 1h 08m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f04e9e253a6f4cc9a3cebb01e5ed0140 : FAILURE in 1h 12m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/881feb1123074ad29b6b1f860b5b74f6 : FAILURE in 30m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c6a5cfde52c145a784ebeda5d5bc019a : FAILURE in 28m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a1a2a56ca7464f4892db4f429f0df5b9 : TIMED_OUT in 2h 03m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0947c570f9052de23841018bc620470ae9f4826d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:21:36.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/mistral is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c2ebd5ec4de6ad3f66b6ae3fb8aa180b93e4d729","revisions":{"9b6fbf01258f5365925ea6a2a97f5ecb2f5673ef":{"kind":"REWORK","_number":1,"created":"2021-05-27 09:14:06.000000000","uploader":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"ref":"refs/changes/26/793326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/26/793326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/26/793326/1"}}},"commit":{"parents":[{"commit":"6e9c47cf03bfa21c4457325a8244969460c39d14","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9c47cf03bfa21c4457325a8244969460c39d14"}]}],"author":{"name":"shubhamdang","email":"shubham@voereir.com","date":"2021-05-27 09:12:56.000000000","tz":330},"committer":{"name":"shubhamdang","email":"shubham@voereir.com","date":"2021-05-27 09:12:56.000000000","tz":330},"subject":"Added sorting and marker for working of pagination of mistral action.","message":"Added sorting and marker for working of pagination of mistral action.\n\nChange-Id: I9bb15642585af22ad354dd4f8f2253a47129639e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b6fbf01258f5365925ea6a2a97f5ecb2f5673ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b6fbf01258f5365925ea6a2a97f5ecb2f5673ef"}]},"branch":"refs/heads/stable/victoria"},"b027de11ce968c5a614016c113316dc0683407c2":{"kind":"REWORK","_number":2,"created":"2021-05-27 12:27:22.000000000","uploader":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"ref":"refs/changes/26/793326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/26/793326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/26/793326/2"}}},"commit":{"parents":[{"commit":"6e9c47cf03bfa21c4457325a8244969460c39d14","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9c47cf03bfa21c4457325a8244969460c39d14"}]}],"author":{"name":"shubhamdang","email":"shubham@voereir.com","date":"2021-05-27 09:12:56.000000000","tz":330},"committer":{"name":"shubhamdang","email":"shubham@voereir.com","date":"2021-05-27 12:26:59.000000000","tz":330},"subject":"Added sorting and marker for working of pagination of mistral action.","message":"Added sorting and marker for working of pagination of mistral action.\n\nChange-Id: I9bb15642585af22ad354dd4f8f2253a47129639e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b027de11ce968c5a614016c113316dc0683407c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b027de11ce968c5a614016c113316dc0683407c2"}]},"branch":"refs/heads/stable/victoria"},"c2ebd5ec4de6ad3f66b6ae3fb8aa180b93e4d729":{"kind":"REWORK","_number":3,"created":"2021-05-27 12:39:37.000000000","uploader":{"_account_id":33128,"name":"Shubham Dang","email":"shubham@voereir.com","username":"shudan"},"ref":"refs/changes/26/793326/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/26/793326/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/793326/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/26/793326/3"}}},"commit":{"parents":[{"commit":"6e9c47cf03bfa21c4457325a8244969460c39d14","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9c47cf03bfa21c4457325a8244969460c39d14"}]}],"author":{"name":"shubhamdang","email":"shubham@voereir.com","date":"2021-05-27 09:12:56.000000000","tz":330},"committer":{"name":"shubhamdang","email":"shubham@voereir.com","date":"2021-05-27 12:39:21.000000000","tz":330},"subject":"Added sorting and marker for working of pagination of mistral action.","message":"Added sorting and marker for working of pagination of mistral action.\n\nChange-Id: I9bb15642585af22ad354dd4f8f2253a47129639e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2ebd5ec4de6ad3f66b6ae3fb8aa180b93e4d729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2ebd5ec4de6ad3f66b6ae3fb8aa180b93e4d729"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
