)]}'
{"id":"openstack%2Fmistral~793528","triplet_id":"openstack%2Fmistral~master~Ia4b1eec7a788cb1ede3dc338f7be7e5d8a84069d","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Ia4b1eec7a788cb1ede3dc338f7be7e5d8a84069d","subject":"Remove references to sys.version_info","status":"MERGED","created":"2021-05-28 01:52:51.000000000","updated":"2021-07-10 14:57:20.000000000","submitted":"2021-07-10 14:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793528","meta_rev_id":"59fa5cd082ddbb319ea9e288fdfb3821af4fd067","_number":793528,"virtual_id_number":793528,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-10 14:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2021-07-10 14:57:20.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2021-07-10 14:57:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-10 14:57:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-10 14:11:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 05:22:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-30 11:34:46.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-06-16 13:01:11.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2021-07-09 17:25:08.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"}],"messages":[{"id":"bdd3485073015470fef527bcf0bad0c1e7f0e942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-28 01:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b381a2294fff9adda66d0c9d271dc67ec762891f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 05:22:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b941ac5202a34578896f67b3e2d9d788 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b33d762fcf44572b775e3861adf3691 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55872f5235124195ba20bf026772615d : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6392844c475c4564b54a236c1809b5c1 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bae3a211e94ac695be7a515bcea96c : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1792b1d5362f41d79f07747d0a60ef6d : SUCCESS in 8m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/89b57435529c4239b179bb24ea92cf72 : SUCCESS in 35m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0af0939c30f4e17afa7e20dedb61781 : SUCCESS in 1h 59m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c8ae9192e7f4edcbfd9fce7ab7afb43 : SUCCESS in 39m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5f06a73e9c594628aa022a1391038fae : SUCCESS in 37m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/24b67316879d42178e5fd5c71f4642b6 : SUCCESS in 1h 59m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a6e59b9a49ee4e369f621cd150e69568 : SUCCESS in 30m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e02b501b1abf4546af987861e7f33741 : SUCCESS in 28m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bfb72e6b11d04e0caac941ffbc0703ed : SUCCESS in 1h 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a00dc428dd5a186d64788d5705087e9ab15064d2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-05-30 11:34:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac0e202a360ae7734c2d1b99c032dc51f23c299a","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-06-16 13:01:11.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"a80f8b7303f37f87be62a3f01b07e83c67d500dd","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2021-07-09 17:25:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c854388b5f76adde26a95a45e76ac2d5d6dd5dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 17:25:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be8726ca5fb1edf7a4ecbb698dcd7f5451ff3303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 18:19:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1541b152c44155a3ff74f0055a075b : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bab02420fdf04487a61bde9f2132737b : FAILURE in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5bda25c037c48df8320c1792d6f9dbd : FAILURE in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd3ee70d5ca4388a16cfe0b5c1f4f31 : SUCCESS in 8m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/77528ea1d7cd4feb8441ce12dd29f3cc : SUCCESS in 42m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98797324619d4ef7b3146dac8d138a56 : SUCCESS in 47m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23f98581e1ca440d8409af6d19fb4633 : SUCCESS in 43m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fd2df0df4d2e49c995316aafb849e2e3 : SUCCESS in 43m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ded3d7a84c8b455dbef84ec73dd213b7 : SUCCESS in 42m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b47ffed347234baa903b2fd33dbab491 : FAILURE in 28m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c8034479b89c4b1bbaca43fbdb2473ee : FAILURE in 26m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3d3e8c6e67fe685fa7a6b193d71ef5c6f07589e6","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-10 02:07:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"402fcfefb37dfd5d480b53b18bc51e16675546cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 03:43:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6143005ce2674a71a4d59a691d8a4ea4 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b702a155fe6e4f7d9d7d0fb350db76a8 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17a0536fdc34c5bab14edcee624316e : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22a24e18e0c34a6cbab037a882317dd2 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5718bd1d7a38469cb78cc8d6e2323888 : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de65a7a003914b9c8ab0566cca6e0b16 : SUCCESS in 10m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/45bae622da734b729925bff8d2fc4625 : SUCCESS in 48m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b8169b66cbb4964a1103178250e6625 : SUCCESS in 40m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81381f6f036d449d84bbebf1e3261527 : SUCCESS in 40m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b210205de81741f4afc685cb8f29f2ee : SUCCESS in 44m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/80e1c8d3bbee410b8b522c6fca72d518 : SUCCESS in 40m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0de9ce3f69f44cb1acbbe00f206253ef : SUCCESS in 28m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1ac5c8efbeaf42668c52c6a2801b7302 : SUCCESS in 27m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/184082a629524eeea4754895c64e2e56 : SUCCESS in 1h 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1afb8b131bff96410d269b76eee5541059b84f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 03:44:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6edfc8bf1c2368baa572081f9e2f254cc1ce67ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 04:35:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4154a3906b9242a489d30da35134f5bd : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c543ea1ecc8441d5801a02e324ebec16 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03862aaf6c6a4c7daf2518d37c020211 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7f1e4f047241e6a7bedd6a228fb879 : SUCCESS in 7m 33s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d0477811ecf843ee916da8842be5d0a7 : SUCCESS in 29m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad81be0f535a48bfa626df43ec51c672 : SUCCESS in 37m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928f7dc2ac0746659c98e66ddf866a11 : SUCCESS in 29m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/74e4b19d04614440993ab0c6aed83bf2 : SUCCESS in 36m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f354e280e1914a02ab3c7885b00bdc62 : SUCCESS in 39m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/66a6ae66c5d94ec7bd4721da33597072 : FAILURE in 27m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df8a61f42bf44f219cec29208062e7d6 : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1524b403797f90bb40a645ad3f412e2edbd039ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-10 06:20:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c6c4eb7472e8545fcebb594d618743b9af28e614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 10:06:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2ab24f543145d88f9aed4d030a8d93 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48a03ea887004726ade9d9ee5d7146d3 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d66d474396a84dc8aef319d4f4e10525 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d37cba89ce8d4f8188e6883f5e105ef0 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b943e180f1843f398843a3af24b8dde : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5feafd959614e7d9a7c4ba2b181bbfc : SUCCESS in 15m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b68cbe93cd7044a1888f1804b2f80d11 : SUCCESS in 1h 11m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a0512efb7164a7d829a9c9983800772 : SUCCESS in 43m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45e0a6557528479dac263045c3ac4c36 : SUCCESS in 46m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/14c93d50b66d466db191af0e7170ab84 : SUCCESS in 45m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/686e657fbd1e43489ba2d80d4f0c62ef : SUCCESS in 41m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5c29240e64a847eabada1fa66d43cc4a : SUCCESS in 31m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b97a35eef7634b6eb9bb140a9e2de6e3 : SUCCESS in 29m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9e1ff126f0a54ab6b1974200068e0c49 : SUCCESS in 1h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32b8f8b88421d83d4b8204831b5cdee9a07a9d7d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-10 14:11:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30b35f8cea58570604c76a30ec1b214fadb5a295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 14:12:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81cd20f8e0e484cd195283715d8624caee3d4d23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 14:56:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06780011ba74bf3bde578b1c7b83e8c : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3d0feac7d9a4630bb5a165e07e6449f : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77b8027f93f14adcb1e211a4de84f8ee : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb6b9df237d42139c296602631b9ffd : SUCCESS in 9m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b65856bb90c840a9912a4e83c96a4cba : SUCCESS in 32m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5cfb344a8384e67bb1f2e50a533932a : SUCCESS in 34m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2fa5aaa31ac49a88ce821a25a882da5 : SUCCESS in 37m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/90510929561b42599e449a31f9b3942f : SUCCESS in 32m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/867afce37676484d9737e7112d24248d : SUCCESS in 31m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7b7f9121be594c10b6ae3fdd59dbfd7e : SUCCESS in 30m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/27b749349ba448ddb68244df91b6d5fb : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c27f3e63d78c5ca7bdf6349a39da8add70c96a0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 14:56:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1029f615283f81664f9b5c595582723ecd73c9f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 14:57:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11083f0b05964b93bf0935af9d70e980 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22e822dde29dd68c5868f615629b84b53841c4e6","revisions":{"63d316bdcddeba834e7cdccf2cc270a81ca3aed2":{"kind":"REWORK","_number":1,"created":"2021-05-28 01:52:51.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/28/793528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/793528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/793528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/793528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/793528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/793528/1"}}},"commit":{"parents":[{"commit":"0ee7a3651028a520dc8ef2be14d765502de4d2da","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ee7a3651028a520dc8ef2be14d765502de4d2da"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-28 01:52:20.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-28 01:52:31.000000000","tz":0},"subject":"Remove references to sys.version_info","message":"Remove references to sys.version_info\n\nWe support Python 3.6 as a minimum now, making these checks no-ops.\n\nChange-Id: Ia4b1eec7a788cb1ede3dc338f7be7e5d8a84069d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/63d316bdcddeba834e7cdccf2cc270a81ca3aed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/63d316bdcddeba834e7cdccf2cc270a81ca3aed2"}]},"branch":"refs/heads/master"},"22e822dde29dd68c5868f615629b84b53841c4e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-10 06:20:08.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/28/793528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/793528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/793528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/793528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/793528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/793528/2"}}},"commit":{"parents":[{"commit":"8aac7f807d81973ef1ade92e667c45b93e455d68","subject":"Merge \"Replace deprecated tenacity.Retrying.call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8aac7f807d81973ef1ade92e667c45b93e455d68"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-28 01:52:20.000000000","tz":0},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2021-07-10 06:20:08.000000000","tz":0},"subject":"Remove references to sys.version_info","message":"Remove references to sys.version_info\n\nWe support Python 3.6 as a minimum now, making these checks no-ops.\n\nChange-Id: Ia4b1eec7a788cb1ede3dc338f7be7e5d8a84069d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
