)]}'
{"id":"openstack%2Fmistral~795342","triplet_id":"openstack%2Fmistral~master~Idb28554df350abefb315983fb92efd7584edde1d","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Idb28554df350abefb315983fb92efd7584edde1d","subject":"docs: Update Freenode to OFTC","status":"ABANDONED","created":"2021-06-08 14:05:32.000000000","updated":"2021-06-09 07:43:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"17f3113ef068e6f7c5b3185d349e940185a98188","_number":795342,"virtual_id_number":795342,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-08 14:28:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":-1,"date":"2021-06-09 03:37:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 14:28:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 03:37:44.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"6dd7f809183d24d5af209d1c58b68585af92f571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-08 14:05:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"647ecb00e659e15749b5669e419513d1c391a19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:28:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff0f021b9b942f293e006ea3ea367e0 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1fa731a16e46df911de784e8f6a948 : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f578c8d0adf4a21d9fad42e7763447113272568f","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-06-09 03:37:44.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for patch. But these changes have been taken care in review ID,\nhttps://review.opendev.org/c/openstack/mistral/+/794307\n\nPlease abandon your patch.","accounts_in_message":[],"_revision_number":1},{"id":"17f3113ef068e6f7c5b3185d349e940185a98188","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-09 07:43:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e19fa781b318f834a8a5e6327531e2e2c5749bc","revisions":{"0e19fa781b318f834a8a5e6327531e2e2c5749bc":{"kind":"REWORK","_number":1,"created":"2021-06-08 14:05:32.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/42/795342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/42/795342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/42/795342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/42/795342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/42/795342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/42/795342/1"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-08 14:02:46.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-08 14:05:29.000000000","tz":480},"subject":"docs: Update Freenode to OFTC","message":"docs: Update Freenode to OFTC\n\nsee: http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022718.html\nChange-Id: Idb28554df350abefb315983fb92efd7584edde1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e19fa781b318f834a8a5e6327531e2e2c5749bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e19fa781b318f834a8a5e6327531e2e2c5749bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
