)]}'
{"id":"openstack%2Fmistral~795438","triplet_id":"openstack%2Fmistral~master~I6de0b6c4717e0ca238ab5a35c7c8cb5ab00fed4d","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I6de0b6c4717e0ca238ab5a35c7c8cb5ab00fed4d","subject":"Update IRC network to OFTC","status":"ABANDONED","created":"2021-06-09 02:15:38.000000000","updated":"2021-06-09 07:20:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7934a1745901c8475c3045f0d3f084703a53bb8d","_number":795438,"virtual_id_number":795438,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-09 02:26:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":-1,"date":"2021-06-09 03:38:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 02:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 03:38:44.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"01cdc9ac7e4e100c3e637cae9ab82cf9cc4d7180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-06-09 02:15:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9e4aa0b9dd81e578754516efd7f6c5ff3f10d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 02:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13017a427fab408b88d82bb31f987e0e : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d599a77b050944f9985d0b23b1ac4127 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"dcebe0d2c8d8a33ad47edd4e1f6101c6bc9e608b","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-06-09 03:38:44.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for patch. But these changes have been taken care in review ID,\nhttps://review.opendev.org/c/openstack/mistral/+/794307\n\nPlease abandon your patch.","accounts_in_message":[],"_revision_number":1},{"id":"7934a1745901c8475c3045f0d3f084703a53bb8d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-06-09 07:20:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a839468b8811e2557c98c893614edd4d9b0c8ce","revisions":{"2a839468b8811e2557c98c893614edd4d9b0c8ce":{"kind":"REWORK","_number":1,"created":"2021-06-09 02:15:38.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/38/795438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/38/795438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/795438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/795438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/795438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/38/795438/1"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-06-09 02:14:50.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-06-09 02:14:53.000000000","tz":0},"subject":"Update IRC network to OFTC","message":"Update IRC network to OFTC\n\nChange-Id: I6de0b6c4717e0ca238ab5a35c7c8cb5ab00fed4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a839468b8811e2557c98c893614edd4d9b0c8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a839468b8811e2557c98c893614edd4d9b0c8ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
