)]}'
{"id":"openstack%2Fmistral~799655","triplet_id":"openstack%2Fmistral~master~Ieeb0b127b25d8e95905ca4d81a636f2c2da5f05c","project":"openstack/mistral","branch":"master","topic":"fnmatch","hashtags":[],"change_id":"Ieeb0b127b25d8e95905ca4d81a636f2c2da5f05c","subject":"Replace oslo_utils.fnmatch with fnmatch","status":"MERGED","created":"2021-07-06 12:18:48.000000000","updated":"2021-07-10 01:53:58.000000000","submitted":"2021-07-10 01:52:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799655-fnmatch","meta_rev_id":"e6e287a0a7b2b241643fb0a68bff5c2d5ba16a61","_number":799655,"virtual_id_number":799655,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-10 01:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":2,"date":"2021-07-09 08:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-08 14:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":1,"date":"2021-07-09 08:23:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 14:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 14:15:29.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-07-09 08:23:50.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"}],"messages":[{"id":"3c062fd40f4745bcb884b1378559a74394f27019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-06 12:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85973d3d18e9e2480634bca1bb87706e91118870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-06 12:24:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"49f24c4ee7e6593adf52aadf365eab789fc028fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 14:11:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e843defa38934b2a85ba4c3a5be2f150 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4819f9f34b1b4887b1213801c29a3bb6 : FAILURE in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b013c10070ff4ee687a5522901ed6701 : FAILURE in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71df1275957744c9b25a62364ad860ec : FAILURE in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411dcd3b34934e24aa5f144567a390cf : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f36ae7b9134dcd816f7c064dc711c8 : FAILURE in 9m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/427375f0b807440cb6d64d46ccc750a0 : SUCCESS in 40m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/469a9e5d51854125868d37e21b010498 : SUCCESS in 42m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/054033e83c3d4c2facf8088c1507debb : SUCCESS in 34m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/58ee0fca686d4e98829acd0f11634afa : SUCCESS in 45m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/db2c8e58d2f3462f88606e9a6e6bb28f : SUCCESS in 41m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ed1b3e5cb6204aa99c82f435c448be92 : FAILURE in 28m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1db3547aedbb4e96b8318866a9f9da38 : FAILURE in 26m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d10bc814dcac474e84d714f20cad1a21 : SUCCESS in 1h 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c9d9583f9aea5cd770437a9e23505423f77b82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-08 10:32:17.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a11ab669fdd7f36e21e6606b1d61d01ca8257c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 12:41:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e03bf428167457d96afaacc97ebbc15 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c96c4307ea42c59f4084e7688df5eb : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e40c7af61784d61a88c94b675f1008f : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edc73baa63f041a48957d04c0c9f526e : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8251724f76564e79a669bcf0f9da3c81 : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3709b7ae1ca445779ce20715cb07fb1b : SUCCESS in 10m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ccce5ccdcf5a43af8bfe35f842828350 : SUCCESS in 41m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/311c6acd25d6481a93573b0b0a570b27 : SUCCESS in 33m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a8f7c711e6b40608f2efc9f93e2dc31 : SUCCESS in 32m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/45f343acbff9429ea64f792d0968fbea : SUCCESS in 34m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/26ab832c05654cdba0d68785601dd3cc : SUCCESS in 40m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/722f29c1cd614d54a87d6a917c780d13 : SUCCESS in 37m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cd0d055055264830b4447c68f7be17f7 : SUCCESS in 29m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ef341b6a8c714d73846683100fef2845 : TIMED_OUT in 2h 04m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b2adba885c640e90fcc34267e3612e2618e5237","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-08 14:15:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9be6b62a77ec50114e291f1e8d68deeb3cc3a793","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2021-07-09 08:23:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e99aa6005591897c8a7deed355052c63fc5aabff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 08:24:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"92ce560556293b228082a683b8cd2121f78e820f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:10:33.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"9b28d07181cdf1edb7174fbba014496fd9204f40","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 16:12:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ee57fe62213525b95aa60a5be5122562f40171da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 21:53:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a76880cd744174af5f49ff34e8fa82 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4869e0c9793d4099b6e5269ece7388d1 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d80ec6fb9d8e463788fdb0637d31a7b2 : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f41f6e8c1c54f3d99bb1c968c395b85 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce830aa83e7a4a16aaae10549b217be3 : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a9206fed19745fea057b1336d55da8e : SUCCESS in 9m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2c6a1a82d5e54e0e97be969cb116ab9f : SUCCESS in 35m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/220ee2e5bdc84b399ab814b763136008 : SUCCESS in 33m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb42760090a34a0798874a9684ff2b64 : SUCCESS in 1h 08m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/206b5323650a41d38ba2e6247edce3e0 : SUCCESS in 31m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/168fb7087f8f44d7a684d08b4e3d02cd : SUCCESS in 30m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6ed64cf21f54452d947bb12cb02976ab : SUCCESS in 29m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/74ed7b78b3e242c887b9d7f34ed322e9 : SUCCESS in 28m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b59727929b0246e49bb46e318e3bdbf0 : SUCCESS in 1h 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52dbbef784b99fb68cf2c9912a5904dd34e57686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 00:05:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"39d869ca7ce62ed4cbe8005087f659cd71d363cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 01:52:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0fbca982fd74e0b810b4d7d379aabdb : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bbacc304cb34e57bc7e1309bb72d713 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84756d0fe5a44df9ae1cf5980aadc9e0 : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d107f635c79f412d902b588eac20abb0 : SUCCESS in 10m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7cd5c814fd2b4cfa902a339170c86eb7 : SUCCESS in 33m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b987898e27e8413f8231fa4d1f9a0544 : SUCCESS in 35m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86c1c5c4845e4cfd963b9949830c0c7a : SUCCESS in 52m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/810c45070f924f78b7d64f703343024d : SUCCESS in 53m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/da7c45f9ed0d476aba171f6f838de813 : SUCCESS in 43m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5f2b564973354dab844f6aea7c6f3fa8 : SUCCESS in 29m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c7ba94e655cc401d9d8bad83af6442be : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":3},{"id":"255ee4b11a2c741884a424cd82f0b4781b752c1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 01:52:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e6e287a0a7b2b241643fb0a68bff5c2d5ba16a61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 01:53:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f736f82f6614a22bd25cbe91b1ecf8d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0bba69725e887b837bda3423e23643270d7899e2","revisions":{"77a1a5d6fef44741b8ad8c9e6ef0d3829877907c":{"kind":"REWORK","_number":1,"created":"2021-07-06 12:18:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/799655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/799655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/799655/1"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:18:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:18:45.000000000","tz":540},"subject":"Replace oslo_utils.fnmatch by fnmatch","message":"Replace oslo_utils.fnmatch by fnmatch\n\nThe oslo_utils.fnmatch module was added to solve an issue in py2.7 but\nit is no longer required because py2.7 is no longer supported.\nThe module was deprecated since oslo.utils 4.9.1[1] and the stdlib\u0027s\nfnmatch module should be used instead.\n\n[1] 4c893c92f551c9dd2a7cfbe7ae8171ad8139df0b\n\nChange-Id: Ieeb0b127b25d8e95905ca4d81a636f2c2da5f05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77a1a5d6fef44741b8ad8c9e6ef0d3829877907c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77a1a5d6fef44741b8ad8c9e6ef0d3829877907c"}]},"branch":"refs/heads/master"},"cff5ff65c37093f57d6103bf07a3f453273683ce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-06 12:24:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/799655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/799655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/799655/2"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:18:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:24:21.000000000","tz":540},"subject":"Replace oslo_utils.fnmatch with fnmatch","message":"Replace oslo_utils.fnmatch with fnmatch\n\nThe oslo_utils.fnmatch module was added to solve an issue in py2.7 but\nit is no longer required because py2.7 is no longer supported.\nThe module was deprecated since oslo.utils 4.9.1[1] and the stdlib\u0027s\nfnmatch module should be used instead.\n\n[1] 4c893c92f551c9dd2a7cfbe7ae8171ad8139df0b\n\nChange-Id: Ieeb0b127b25d8e95905ca4d81a636f2c2da5f05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cff5ff65c37093f57d6103bf07a3f453273683ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cff5ff65c37093f57d6103bf07a3f453273683ce"}]},"branch":"refs/heads/master"},"0bba69725e887b837bda3423e23643270d7899e2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-08 10:32:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/799655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/799655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/799655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/799655/3"}}},"commit":{"parents":[{"commit":"de880e801cba89f0d674a2a1f2435fb7d9942670","subject":"Fix unit test failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de880e801cba89f0d674a2a1f2435fb7d9942670"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:18:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-08 10:32:17.000000000","tz":0},"subject":"Replace oslo_utils.fnmatch with fnmatch","message":"Replace oslo_utils.fnmatch with fnmatch\n\nThe oslo_utils.fnmatch module was added to solve an issue in py2.7 but\nit is no longer required because py2.7 is no longer supported.\nThe module was deprecated since oslo.utils 4.9.1[1] and the stdlib\u0027s\nfnmatch module should be used instead.\n\n[1] 4c893c92f551c9dd2a7cfbe7ae8171ad8139df0b\n\nChange-Id: Ieeb0b127b25d8e95905ca4d81a636f2c2da5f05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0bba69725e887b837bda3423e23643270d7899e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0bba69725e887b837bda3423e23643270d7899e2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
