)]}'
{"id":"openstack%2Fmistral~799727","triplet_id":"openstack%2Fmistral~master~Ie080bdcacf9aa2dd2fe0454bcb814285c08bbfa8","project":"openstack/mistral","branch":"master","topic":"ut","hashtags":[],"change_id":"Ie080bdcacf9aa2dd2fe0454bcb814285c08bbfa8","subject":"Fix unit test failures","status":"MERGED","created":"2021-07-07 00:26:20.000000000","updated":"2021-07-10 01:19:22.000000000","submitted":"2021-07-10 01:18:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799727-ut","meta_rev_id":"0ac316a504f2963b77381a049633dfc4833bbd23","_number":799727,"virtual_id_number":799727,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-10 01:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":2,"date":"2021-07-09 08:24:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-07 14:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":1,"date":"2021-07-09 08:24:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 04:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 14:08:19.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-07-09 08:24:24.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"}],"messages":[{"id":"b4cef32a8989b6e650639ecff4715751877894e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 00:26:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"406972b5929c046acadd14370bff1c1e792345b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 02:05:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1101a56cfeaec2ff7e9b32b71625c4455474d76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 02:31:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"70556030ae086d47739099c4b6f174f11eb6e223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 04:22:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6b4f42766e4848bfe2dccf0ee08725 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48c9825ea70a41eebc72ae95f9a19327 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2bed681088c46e1b57c1c8a97de04ce : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df9b912c15b24586acd2543ac6436159 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b6cdb7ce2a4ee7a6221f32898ceca8 : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b4084e8b72f442caf7fbb5d7a1d3eff : SUCCESS in 13m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a5efe18a7ab5493699c6c8b9fdc56817 : SUCCESS in 32m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4db621cb0bac439aae39e31ad7b2e4e9 : SUCCESS in 44m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da4f5d765ee64d6c8831e25a6786ea9a : SUCCESS in 32m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8b2deebc0b0340ab87e8842fe9bfe912 : SUCCESS in 41m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8912621940d24f6b9d7f4fac838db343 : SUCCESS in 50m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ba7c7149cde94179a141916af2270add : SUCCESS in 28m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/598068aaa88a41cea3ec5c9ed90ee89a : FAILURE in 26m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/28aed174ab154439baff63df34ebf780 : FAILURE in 1h 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"82e367679fba9ebe6ee1e424da972dad85922f84","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 07:12:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"70aee42472103824ad93f92b46d041e38ef53895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 09:09:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67de8623d12d4cdaa7331590787f939a : SUCCESS in 9m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbad749f4f744e818ee9f98fc75a2873 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42b25de6bd534110b41a4a6c75dcd2f4 : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2210465507814f6e849a1ed5d93ec3f9 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851474287f2b472aa6ec956423960778 : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ffe649c6f834267989064504d17f059 : SUCCESS in 9m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3149683578a848d798b6956b8c3de629 : SUCCESS in 45m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c32048ce7b549a491c51d5a6b83f10a : SUCCESS in 44m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a46944621194186893bc5c9bd522919 : SUCCESS in 42m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/93b64237d880435ab0488aa03685b92e : SUCCESS in 41m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6f51ac0b4d834bf6aa0062a09067b902 : SUCCESS in 48m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bac3ce51b9aa4176bc79b7f567a5ddf3 : SUCCESS in 29m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/062b07888b8f450cb0ed2c122da96f04 : SUCCESS in 28m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4489853af018400e916b71190f3a3385 : SUCCESS in 1h 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58430f5035846f02b7931d8d5c2e295b83d5ef27","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-07 14:08:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e7362134876691706c787e19301b78d4ceac1e72","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2021-07-09 08:24:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8cf51a4fe77ec4fee1dbc15c26102a69ac44c704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 08:24:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e32b798ca61c96d6896e1681431e626e54555a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:10:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9517e45101849848284c2592c4e0513 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c43fdd08c94fffa7958bf75bc9d08b : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cfbc625843b4062838e4839f7898995 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b552fb0fa3b9494a9b0f30312c630d0a : SUCCESS in 9m 30s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a80ae8985ea246418da42908acbdc301 : SUCCESS in 39m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32bcdd5953bd4f13adac099008ce8e47 : SUCCESS in 41m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48eb6b59215944b2aa1970f54af12a30 : SUCCESS in 44m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/161f4962713b4335a239591e8293bf70 : SUCCESS in 42m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e65c426f49d642c69708190bb54a21a1 : SUCCESS in 44m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3eaa19116a6a439ab6793c987c105dfe : FAILURE in 29m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bc5c355ae11d4ef98a0c773a5b716b3b : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ccd37bffbfcfbefcdf48d21d86f56fcb6e220fed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 16:12:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"739c789a5775258aecfbc77ac5b5893eb60e48b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 18:23:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12637a0f2244bb3ae4f110ff0aa1ca7 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83af501bfe0947f7899830f164eec7d3 : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a24486d976e44fbb6b41ff8ccced845 : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c509a67332034c1099a1ae4f7b917931 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7322b064111b447981983fe53fdc39ce : SUCCESS in 9m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baada2832cae45ba9926deb707468c30 : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c6f2fa52a9cb43ce8fc673242a369eb6 : SUCCESS in 35m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63c8bb69f9534140aa21154be5df683d : SUCCESS in 37m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58e5705d8db74646bd9333b7754d2298 : SUCCESS in 35m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e0e9717b18a8425d9a155551c411b393 : SUCCESS in 39m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/26ba58ae8d2b43daaad7f6945945ae2a : SUCCESS in 37m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/80f4489e2336402087321ea340e70a92 : SUCCESS in 29m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/08a5e97a02a14455a4ee661c2405f8b2 : FAILURE in 30m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0fd9cb867a2a408f8ca1ea4df4e968db : TIMED_OUT in 2h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad3c7a48317ca3b374df9d03bcc211e6b0d17eb3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 22:11:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"965df5182789e4f9411cded493ce0704aaf79d1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 00:05:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d502a208e445569326a12c4bfff035 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b3863b7bc884d26a2586d43f29ef664 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32ec67de7a9f4ca0b825ab0a8b6ba52d : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/902077f49efd47faa405392c022afe9d : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff22cba4321544a6a41fd4d286b8b0cf : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ed9861538a49f5a10dd5f5d06add7b : SUCCESS in 9m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a36951e25d6147559e428175954859b7 : SUCCESS in 39m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b1150c142584de0ac095ea1eb1d4ba7 : SUCCESS in 36m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82e643ebf17948e2976735ad4021bfe7 : SUCCESS in 41m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/578673e7831a485e8dfa2e9208cbc13b : SUCCESS in 41m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/50f93dc8d23144eb9b422edb9085175d : SUCCESS in 39m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ade824fa90ae420b82e732f42addcfe6 : SUCCESS in 29m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b5d4787b0d494044b73f2d17dc5f090c : SUCCESS in 27m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b3799ac15c95430582d7ceb43b777601 : SUCCESS in 1h 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f6641bc9c86ab6ee9d37a378fd7651cf7b373d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 00:05:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"902a9bf4bac7f5eac064fda4b018b058819d3fca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 01:18:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/620d21b1fb174adeb2b3e2b262dae784 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/873fc052849c479db268439f6a105332 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f25ae977f744f94a27759e5d281a10e : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2ff2f69dcb42a1b466cfdbd95e2832 : SUCCESS in 7m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1c6e5ef76b7a4c99a749bdd4c540a063 : SUCCESS in 42m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88e2f5c5686e48c6820d5a1f626d2981 : SUCCESS in 36m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a058ec8c12114a5fbba569dcb5c3aaa5 : SUCCESS in 36m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/384c136302e14ee788308cd19274f1c7 : SUCCESS in 35m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6ea2b005182e4998ada9a4670a22f07a : SUCCESS in 39m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/be0818b865b64b0e8a41669e844eebb8 : SUCCESS in 30m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/147ef3a79d324bd4bd85a901eee5dc0d : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":3},{"id":"55dd29b80131b6c7f170ada96555bd05c2843a39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 01:18:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0ac316a504f2963b77381a049633dfc4833bbd23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 01:19:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fead4934c640049adc90864722c1e4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"de880e801cba89f0d674a2a1f2435fb7d9942670","revisions":{"fedcc5cfc836189faed723b384e0498e9c2026d5":{"kind":"REWORK","_number":1,"created":"2021-07-07 00:26:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/799727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/27/799727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/27/799727/1"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 00:26:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 00:26:08.000000000","tz":540},"subject":"WIP: Fix unit test failures","message":"WIP: Fix unit test failures\n\nChange-Id: Ie080bdcacf9aa2dd2fe0454bcb814285c08bbfa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fedcc5cfc836189faed723b384e0498e9c2026d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fedcc5cfc836189faed723b384e0498e9c2026d5"}]},"branch":"refs/heads/master"},"777ddadbe9f6608001475fdaf9a0b4f581313b47":{"kind":"REWORK","_number":2,"created":"2021-07-07 02:05:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/799727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/27/799727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/27/799727/2"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 00:26:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 02:05:01.000000000","tz":540},"subject":"WIP: Fix unit test failures","message":"WIP: Fix unit test failures\n\nChange-Id: Ie080bdcacf9aa2dd2fe0454bcb814285c08bbfa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/777ddadbe9f6608001475fdaf9a0b4f581313b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/777ddadbe9f6608001475fdaf9a0b4f581313b47"}]},"branch":"refs/heads/master"},"de880e801cba89f0d674a2a1f2435fb7d9942670":{"kind":"REWORK","_number":3,"created":"2021-07-07 02:31:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/799727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/27/799727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/799727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/27/799727/3"}}},"commit":{"parents":[{"commit":"c63e6e086ea119aa51d4f8b86ba1096a19f86b4a","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c63e6e086ea119aa51d4f8b86ba1096a19f86b4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 00:26:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-07 02:29:17.000000000","tz":540},"subject":"Fix unit test failures","message":"Fix unit test failures\n\nThis change fixes the following two patterns of errors detected in\ngate.\n\n1)\nsqlalchemy.exc.ArgumentError: Textual column expression \u0027...\u0027 should be\nexplicitly declared with text(\u0027...\u0027), or use column(\u0027...\u0027) for more\nspecificity\n\n2)\ntesttools.matchers._impl.MismatchError: \u0027(\u0027...\u0027,)\u0027 is not an instance\nof tuple\n\nChange-Id: Ie080bdcacf9aa2dd2fe0454bcb814285c08bbfa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de880e801cba89f0d674a2a1f2435fb7d9942670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de880e801cba89f0d674a2a1f2435fb7d9942670"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
