)]}'
{"id":"openstack%2Fmistral~800405","triplet_id":"openstack%2Fmistral~master~I9bf3de8a5ae6a8c9346285147732b773a3667f7e","project":"openstack/mistral","branch":"master","topic":"healthcheck","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-14 00:35:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9bf3de8a5ae6a8c9346285147732b773a3667f7e","subject":"Add optional healthcheck middleware","status":"MERGED","created":"2021-07-12 04:12:18.000000000","updated":"2023-12-14 00:36:57.000000000","submitted":"2023-12-14 00:35:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"800405-healthcheck","meta_rev_id":"c0a21f1f7cd1eb7d98d6c54f88396292e5a239dd","_number":800405,"virtual_id_number":800405,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-14 00:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-03 04:09:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":2,"date":"2023-12-13 22:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":1,"date":"2023-12-13 22:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 06:23:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 15:47:11.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2023-02-03 13:37:17.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-12-13 22:48:57.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"1ce8f5d50c32cab84eba22caa810146740a2e8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 04:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09eccd2bbfd678dabba2fa4f97424a0a5d37ab6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 04:15:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6deb117ed635482497c27c26883f2dc6b39e2ca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 04:22:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31379ba6c2fe0e33960987ded7062f4476496537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 06:23:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df810f2e06984e598e5bbb3eb9baa360 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2349c2688c240c2ada4e90c0d853a56 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9ab8be55124ba4a85fb661ebeac82d : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85a4c40a94f545c2ad0bd07b9a65704d : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4919c32a14ee4efe9a94545345109c89 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a190d5481f7442cc82cf3e83055af17b : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17740283fc214e579fb6d6387495b4a5 : SUCCESS in 9m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b4bc7778e5c74656afa4f0d2fa091d6a : SUCCESS in 44m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12bdebdcb8ad446891e74e07dc89e0eb : SUCCESS in 44m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/269088775a874cb1a4dcf82a0e56339c : SUCCESS in 47m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a5b979b4f01e483e9658d71baaedefe2 : SUCCESS in 37m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f90acfa9b85c4e5fb2053feb781a1fac : SUCCESS in 39m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8f4e0fac6ba04a718b29001c39ea2eb5 : SUCCESS in 29m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9f8c7c1effa84d8fa01b20421438cdea : SUCCESS in 27m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ccd4c71f9b0e43ff89ee37ad82a1d1ef : SUCCESS in 2h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"572db56008a34189975655394c495b0853ddee7c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-08-03 15:47:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7f53a2f66acf964c47a055ece71ad518b232d7b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-21 02:05:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"373c45d99f9b3674936eabbb199dd68de9e78908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 04:00:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b21f57806dc4606ab1bef1f35e528b6 : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053628c1ccbe48a28fef0e56c5565081 : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/096adc19641d41d4abf8a6f61c3db251 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b48a1fba2634052861f83801a29485c : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce89b7090f9498ea1485af6b2c07f0c : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/858699896c1946f1a4b16c4c00b2f355 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/564b15cf0151498793567d2fbe7b91dd : SUCCESS in 9m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a0dc36f7913143658a06544d7ee44911 : SUCCESS in 36m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e82cf67553b4da0b750bf250856eda9 : SUCCESS in 1h 08m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c74dfa860f9f4de99151493801b10df2 : SUCCESS in 36m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3a1b749af3a847ac825bca70ece8bce7 : SUCCESS in 35m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/502e048670914b78a7d4e92d2818edd2 : SUCCESS in 36m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/42f1161de8d244f2b2cf9338acca1659 : SUCCESS in 29m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/927060c6860b4f288b6f89cb48571046 : SUCCESS in 28m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ec48651fac184d7093fc8fc27ed5e2d3 : SUCCESS in 1h 51m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"54dacb4e99f3904b01238df96c528d78e0086b42","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-09-05 08:45:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6a1c25fba22ce553815564493f95dec2253929c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-03 04:09:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"0ba22e7f915b7d52cedaec6331dee9027ffc9655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 05:48:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a5e0209f09c46f28bf192d5ac2741a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed3bce577754b41885f7eb214f51963 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c913fe4c9fe4d7ca8f716b968dc75c0 : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7859712bf4ba4e8c81e9a9e9d825a49b : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1a1051b1794de3a3641897dc6591e6 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c1d8eaa923c4ec087a8ac18ce450e5d : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dbaf3e4116f47b2b3ffb91287d52fcd : SUCCESS in 11m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b6d3862461204a07a952b94d2d32c25b : SUCCESS in 36m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fe6b7a6b643414cbbfae7935142505d : SUCCESS in 36m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1acae27180d84163afcd02296d80aa40 : SUCCESS in 42m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f11305ca801a411e88002c71e0152149 : SUCCESS in 36m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/11ef949823144191a05a29c144461fc2 : SUCCESS in 42m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a9b4b600994e4253829b1caecdef632a : SUCCESS in 33m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/85388df78fd841a79202ec68ffe7c1c3 : SUCCESS in 37m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/978fa2b455b24e02b2a62367d7812f72 : FAILURE in 1h 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2ad8f516f274c90a735cc2808ea11b0d63db0617","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-03 13:37:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ebf5b524bad2354302702436f7050eb650f05ab4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-03 13:38:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfa61bceb25776dae3538ab03fd39bce2e5c9442","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-13 22:48:57.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5546683739623ac6d2564b1271ced3d4cb7626e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:49:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1056e7477221c6ab71f8a3ba68e47e674ae10814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 00:35:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5be107fc6c3480c8b30ae7bef1f1b01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f4dd79a7fd4e5a8d0dd6b2532573db : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b90b772d09cb42f3911334133798a3f5 : SUCCESS in 9m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31e46ce68f6143a4bb0721566327fa8b : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b0082900ed422ab75b1f6758a26672 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0063d714341410a91dbb5fceb259552 : SUCCESS in 3m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5aad4f10531e460b8d0b4603926ec070 : SUCCESS in 27m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59ef6f63f79b49bd85271284e8e7eba3 : SUCCESS in 29m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3521707a219f4ef48eabebafa519b5a2 : SUCCESS in 48m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ee063acbf17840ac9f6ad6bfbe533cb3 : SUCCESS in 46m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8864487432f9496db8e68627c9a57927 : SUCCESS in 27m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c87aaa572d174991a4c69e85acf980c6 : SUCCESS in 56m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3c36860102b44140822d618e157f6768 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":5},{"id":"12667eab7df10f51cdcd6af049a33986848ee364","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 00:35:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c0a21f1f7cd1eb7d98d6c54f88396292e5a239dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 00:36:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6e5e95abc5c42c4a61ac1e60466392b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af36c13db576451cb370a696d4365e8d : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67a748ec8c714e14a9e9851821bc4695 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f17dd7a04bd384aaef1bb679dbab0b9749294ea7","revisions":{"3c92efa3955ce3268e94298362ceada997b93d12":{"kind":"REWORK","_number":1,"created":"2021-07-12 04:12:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/800405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/800405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/800405/1"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:06:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:10:37.000000000","tz":540},"subject":"Add support for healthcheck middleware","message":"Add support for healthcheck middleware\n\nThis change introduces a new option, [healthcheck] enabled, so that\nusers can enable healthcheck middleware in oslo middelware repo.\n\nThis change is created based on the same change proposed to ironic[1].\n\n[1] 6f439414bdcef9fc02f844f475ec798d48d42558\n\nCo-Authored-By: Jim Rollenhagen \u003cjim@jimrollenhagen.com\u003e\nChange-Id: I9bf3de8a5ae6a8c9346285147732b773a3667f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c92efa3955ce3268e94298362ceada997b93d12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c92efa3955ce3268e94298362ceada997b93d12"}]},"branch":"refs/heads/master"},"bbdd1f840c548b6d7b27a4f2a28b58feca88ced6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-12 04:15:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/800405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/800405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/800405/2"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:06:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:12:46.000000000","tz":540},"subject":"Add optional healthcheck middleware","message":"Add optional healthcheck middleware\n\nThis change introduces a new option, [healthcheck] enabled, which\nenables the healthcheck middleware in mistral-api pipeline.\n\nThis middleware allows status check at /healthcheck path, which is\nuseful for load balancers or any monitoring services to validate health\nof its backend services.\n\nThis change is created based on the same change proposed to ironic[1].\n\n[1] 6f439414bdcef9fc02f844f475ec798d48d42558\n\nCo-Authored-By: Jim Rollenhagen \u003cjim@jimrollenhagen.com\u003e\nChange-Id: I9bf3de8a5ae6a8c9346285147732b773a3667f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbdd1f840c548b6d7b27a4f2a28b58feca88ced6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbdd1f840c548b6d7b27a4f2a28b58feca88ced6"}]},"branch":"refs/heads/master"},"ad22e21cc7b6248260e1f769dfd8b959880475aa":{"kind":"REWORK","_number":3,"created":"2021-07-12 04:22:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/800405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/800405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/800405/3"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:06:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:22:32.000000000","tz":540},"subject":"Add optional healthcheck middleware","message":"Add optional healthcheck middleware\n\nThis change introduces a new option, [healthcheck] enabled, which\nenables the healthcheck middleware in mistral-api pipeline.\n\nThis middleware allows status check at /healthcheck path, which is\nuseful for load balancers or any monitoring services to validate health\nof its backend services.\n\nThis change is created based on the same change proposed to ironic[1].\n\n[1] 6f439414bdcef9fc02f844f475ec798d48d42558\n\nCo-Authored-By: Jim Rollenhagen \u003cjim@jimrollenhagen.com\u003e\nChange-Id: I9bf3de8a5ae6a8c9346285147732b773a3667f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad22e21cc7b6248260e1f769dfd8b959880475aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad22e21cc7b6248260e1f769dfd8b959880475aa"}]},"branch":"refs/heads/master"},"e2ebf3cbb6e3a207b53a36d3fe32a74e0e605343":{"kind":"REWORK","_number":4,"created":"2021-08-21 02:05:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/800405/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/800405/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/800405/4"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:06:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 02:05:46.000000000","tz":540},"subject":"Add optional healthcheck middleware","message":"Add optional healthcheck middleware\n\nThis change introduces a new option, [healthcheck] enabled, which\nenables the healthcheck middleware in mistral-api pipeline.\n\nThis middleware allows status check at /healthcheck path, which is\nuseful for load balancers or any monitoring services to validate health\nof its backend services.\n\nThis change is created based on the same change proposed to ironic[1].\n\n[1] 6f439414bdcef9fc02f844f475ec798d48d42558\n\nCo-Authored-By: Jim Rollenhagen \u003cjim@jimrollenhagen.com\u003e\nChange-Id: I9bf3de8a5ae6a8c9346285147732b773a3667f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2ebf3cbb6e3a207b53a36d3fe32a74e0e605343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2ebf3cbb6e3a207b53a36d3fe32a74e0e605343"}]},"branch":"refs/heads/master"},"f17dd7a04bd384aaef1bb679dbab0b9749294ea7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-03 04:09:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/800405/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/05/800405/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/05/800405/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/05/800405/5"}}},"commit":{"parents":[{"commit":"7c35734300d5d700928ebdf2315e392004121f39","subject":"Merge mistral tasks data to execution context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c35734300d5d700928ebdf2315e392004121f39"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 04:06:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-03 04:09:13.000000000","tz":0},"subject":"Add optional healthcheck middleware","message":"Add optional healthcheck middleware\n\nThis change introduces a new option, [healthcheck] enabled, which\nenables the healthcheck middleware in mistral-api pipeline.\n\nThis middleware allows status check at /healthcheck path, which is\nuseful for load balancers or any monitoring services to validate health\nof its backend services.\n\nThis change is created based on the same change proposed to ironic[1].\n\n[1] 6f439414bdcef9fc02f844f475ec798d48d42558\n\nCo-Authored-By: Jim Rollenhagen \u003cjim@jimrollenhagen.com\u003e\nChange-Id: I9bf3de8a5ae6a8c9346285147732b773a3667f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f17dd7a04bd384aaef1bb679dbab0b9749294ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f17dd7a04bd384aaef1bb679dbab0b9749294ea7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
