)]}'
{"id":"openstack%2Fmistral~801078","triplet_id":"openstack%2Fmistral~master~Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0","project":"openstack/mistral","branch":"master","topic":"bug/1936667","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2022-03-04 17:04:22.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-03-02 08:08:42.000000000","reason":"Adriano Petrich replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-04 17:04:22.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-03-02 08:33:49.000000000","reason":"Eyal replied on the change"}},"hashtags":[],"change_id":"Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0","subject":"Replace deprecated import of ABCs from collections","status":"MERGED","created":"2021-07-16 15:14:44.000000000","updated":"2022-03-04 17:14:15.000000000","submitted":"2022-03-04 17:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801078-bug/1936667","meta_rev_id":"09a3204a439c589c415fed5a84bf1989a8b70820","_number":801078,"virtual_id_number":801078,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-04 17:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2022-03-04 17:14:15.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2022-03-04 17:14:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-04 16:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-04 16:19:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 16:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 07:24:53.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2022-03-02 00:22:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-03-02 00:22:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-03-02 08:26:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"2b973385599e9137fbd978ccdc0cdcffa439c401","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 15:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f1bcaad5b10567dff26ee4575e294ce9d801c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 15:18:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a536de5d2ff2d121baf02887ebe7dc84d56ca488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 15:21:45.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a6ce08e688f1c27466b293bdc029953fc39fa4c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:13:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c8bb0b8a8045e28c1ed6cb5c75d813 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb844ddd2084fe5bc52855ab7ffc9ce : FAILURE in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d4d61adce242dcbd8e555a03aa0265 : FAILURE in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f8a9ef60e0448fa952ec1affbf68bdd : FAILURE in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6769614cec14d2fa017b983379d0aaf : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a78de21bc8840d7a08d08e8164e22fb : FAILURE in 9m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/87d596976dde431e8ce790dc79225d9c : SUCCESS in 45m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc9f2481eb6d49b2bca903b6e94dd103 : SUCCESS in 41m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32340bbe0fc54f949b76cb7474fc78fe : SUCCESS in 41m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/20ce4e307837466bbf286d190f9d5ef5 : SUCCESS in 37m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/16e195fe6e2e4c718be1ed81f5c20882 : SUCCESS in 43m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8a56bd70400044959b20b1924ed222bf : FAILURE in 29m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e458c5316cdf4e2e99b201952956475e : FAILURE in 27m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/985e0db2f8d04173b5a00c1fa99b87ab : FAILURE in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a3398092df241695cfb9931f944d617f128b4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-17 03:46:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ce3dd4f4c84a7510c4c9bcbe262721ee11be2133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 05:23:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa724b80b7384d00a0d32058ee2942ba : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/153c0ad0da3349099b3c6f1ab5497d85 : FAILURE in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7060cc01c3474089b90b34ee0c945790 : FAILURE in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d31e51036a94c4498317ba2b6a93ce4 : FAILURE in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01449b1f38ac4ed18999e6414b4521c4 : SUCCESS in 12m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e8f9efe2de74a57b93539b67eec5e12 : FAILURE in 9m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/253560a1e1174cad9bf1c22d9f247e16 : SUCCESS in 51m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beb2accc02a94f8fa73a43a220dc0f81 : SUCCESS in 39m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/affb56b0fd6c42cbb0698d4da063b7f2 : SUCCESS in 38m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/18c63a4fa9df428e856dc60eb1d59115 : SUCCESS in 39m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/85632248e897476fb17999e9944cab3f : SUCCESS in 35m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ccb69396e4f84e38863a3222f49af72b : FAILURE in 29m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0ebd59d2f6094b7faad795e7bb349b2d : FAILURE in 27m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2467e16c53434f55930cc3a153090eeb : SUCCESS in 1h 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6de7f1d29c68d81382c113d5cb83e18eb0dadd02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 00:34:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"eb9dd59c602287a9f3fb952636c8133ca6e287c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 00:39:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"b85b1c6a04f7630b877c7de2b7d812773fea5e7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 02:21:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ac3cf4eb43472abc48eff31afb4d5f : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89ad906c557f44bd86fd68a26657c5f5 : FAILURE in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/099cf17fb3ee4a149c7a3dc206651f13 : FAILURE in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccce29fa72f748c3b06cf140c36337db : FAILURE in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076bd4d01cec44a9bb31194db55e91c0 : SUCCESS in 8m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4158e28968cf4d85be60e7b2f525e4fd : FAILURE in 8m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7fde38e091dd4134a0ddcd58f06df6b6 : SUCCESS in 39m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dba08af38ff74f718a25ff0842719307 : SUCCESS in 39m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07b73960f2ae4479b845f30775e2076e : SUCCESS in 39m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5578fe5081f3414cbbec0467cdbb6cde : SUCCESS in 47m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dd5c11267ca845d8ac28bcbc26cf948f : SUCCESS in 35m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/70a3002922164ec1bdd211e967d3cf1b : FAILURE in 29m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/807da1b4a659482ba7dfd5f91c741fdd : FAILURE in 27m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/626d80ba6b804da8ad0f8e1e1853e8c3 : SUCCESS in 1h 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"874facda7a1bcd7ffe018754f8e2e94d03d0e429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 02:32:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"2972d0d04e42e1ef924edb826608974928a5c80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 06:46:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14499a3728bb467eb45f837cbbf78140 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3df02cc2144a8da0b503190ed0c35f : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/071c019aedb2429eb5f5620dc454e5fb : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cef5a3507d814663a76252a6a7795443 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876a6388c6414e28a5c4980b789fec18 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1ee5ded18b348bda5a22b384c2d5195 : SUCCESS in 8m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/df3a7f127f2044c28959602bc58fe9cc : SUCCESS in 40m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/326fa321c5d642658666c92308813f1c : SUCCESS in 43m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6fcdb3f78f4b7f9b0ec863fe2805f1 : SUCCESS in 43m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2a06a56ff94c4a1c932f4ec9a800aa65 : SUCCESS in 41m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b32f84b1a9f846578ac82b7e45a2dfec : SUCCESS in 40m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e268c79b3039485e858636773de35343 : FAILURE in 28m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2f6181163b2f4557ae70d44068eb5308 : SUCCESS in 27m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/49c48420d868437d8738953601055771 : TIMED_OUT in 2h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4d0d5dd913903e9b3a1cc378b0734fcf591ced6f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-03 14:52:16.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"780f30e6a1da4437b1063f86ee8a4046e5c35e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 17:01:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c454adbf3611437b8cdd40d601147ca7 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e301d463e3604e2ca8725508b0a49608 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b586a7d07332498b9419644591956f6e : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57507326c49a43aa843b994b26e1d5d3 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed98bbf4650b4bb7832c76dcad1bba56 : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d101e1d54841449ea11ac3cb102cf3 : SUCCESS in 10m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e92fc5cf04cc4798b28a00946a906b71 : SUCCESS in 56m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1476ccb3eb9346cda6c4954ecd5f22e0 : SUCCESS in 36m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e4cddd3f3ad44e1b4b120205ccbb736 : SUCCESS in 37m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/258581059555400194509d060e909762 : SUCCESS in 34m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2db9994513dd4cef9c3fe7f29ce0b437 : SUCCESS in 51m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7a5bd3f6951c4e889eb11fab93ac11b1 : SUCCESS in 30m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/157263cce9ce4f3a92e2d0f05c65a771 : SUCCESS in 30m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6058bd6bed5f40dd99534b5254b54be7 : TIMED_OUT in 2h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5250581b61828393349e9819e951f26110355700","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-08-19 07:24:53.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"65409f74f6f5713f3e049cd6cb8b486e6956c14b","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-03-02 08:08:42.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"9d8bcba5147e4c354b05c60f93a06a5e91b69f7e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-02 08:33:49.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"12300d0ab7c0c517d16152d5d6937eba0e49d2ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:34:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3d494a3484a96c20f931235303bb4cbe74514532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:28:45.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47584756957d494ca1a5efb3e49d152b : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9989bc2332ac44aebcb8c1bc4d873059 : FAILURE in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c17c016e90b844c79d8b45c12e2b9cfb : FAILURE in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5e4cd4eaea45f08073f6d1e2634b85 : SUCCESS in 9m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b2f8bba25a0342fc864069e0bcd180a4 : SUCCESS in 51m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fb43f41cd504a86aea61a02f5a066ee : SUCCESS in 49m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4430bd0fc9a2446a8c52ec74c67a8f77 : SUCCESS in 49m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bd031c09c1174c26a25f33ea55022053 : SUCCESS in 51m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/59904b310c0d404e84d170c200513be7 : SUCCESS in 51m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b2c21840ca7b4041884e384bdb1c0994 : FAILURE in 33m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/84048ccd06ba4cebb599ff8ad2a2d60d : FAILURE in 29m 44s","accounts_in_message":[],"_revision_number":7},{"id":"e9e8d124d7350bac56aa56126fdda28fd6f3862e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 09:29:55.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"d1385f63b7a90c5e2a2a7bbf41786591c1e23fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 09:30:29.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"1a5431267953d23109afa17ec12a7730e20f66a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:47:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32307cd22e3a4b9d9fdea3825b19c168 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5edadc98dfa84e33b81198e069707520 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52ff62e164d4459ca16549f55e63fe8d : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2bddb1ba8704c44ad358dfc9848f3d6 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4bf2b6474fd413eae9fecfb538f5359 : SUCCESS in 10m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a951e8aa2624b5ab2f2f8a44fcef1d1 : SUCCESS in 11m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/32455a5af79042e6aed6c6bb22e253c5 : SUCCESS in 51m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac8c80af23c64e6f9652970e41ec9580 : SUCCESS in 49m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c31668a7bf6b44ee9bec4d5f7d559823 : SUCCESS in 54m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5801637d190f40139b826792dceca7fe : FAILURE in 2h 11m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b32bdb744e6343b99fb14fa8476d22eb : SUCCESS in 55m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3207a369c6a6445d9c9878c4fe1ea545 : FAILURE in 30m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/667c27b7deab488ba939eb5fe373ee5b : SUCCESS in 29m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/96b79d04304e420e96c9e07846166ea6 : TIMED_OUT in 2h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"474472e85a8a2998e158a43e487a81f6ac13b483","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-04 14:07:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b7a6d8575bc80f1194a7004cb2c1f1162e89366e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 16:18:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74cacce5be24c5aa2888c3cd5d27459 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5945e1844b1445bbb47781f0ca929161 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5472cb90964c412082d6fc2d0a2d6e6b : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/032322543f704e6f9d60078ed3bc36c4 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d840ddab8e0479fba17942eddf61ae3 : SUCCESS in 10m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f860887af4494be18a33a83453244e17 : SUCCESS in 11m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9b429ddc4d824f31b77a1eb3a6ba389a : SUCCESS in 39m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97eb0c2612ef4cf3a237f5415619d3e5 : SUCCESS in 44m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/917d63918ddf432d8902a8443aca1a79 : SUCCESS in 43m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a92b752a264f41d8bd729b0b80c62221 : SUCCESS in 42m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bf22b9d29de64a89a8cdea4c74035620 : SUCCESS in 39m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5b26197906324b82897e35f56d7e324d : SUCCESS in 34m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a19abb70470a488c8c29141c630ae0cf : SUCCESS in 28m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/196a9322b0a940faaa3077455dada47f : SUCCESS in 2h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"21f35df750a08eab6e24fa77514af24fe11767f4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-04 16:19:40.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"335b1bd0dd64d7b8bd685fe70a6cdc1f57a71a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 16:20:38.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"47051dbe0e32be1968da8667603eafe58e4ff383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 17:04:22.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745307fa0c0d4f26a9413ac18e15a8a8 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/816db3cdf341499db3123abc8256e77a : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee3bc52e456d4c1d85c1347f48e8f42d : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b64aa23d394771be71970246fcccdb : SUCCESS in 9m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6fb0a822cdc1421d9d743aa671a0e548 : SUCCESS in 38m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c18b3e4bc0348b3b576d9d9ef3e2c94 : SUCCESS in 40m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/700e375e42a6426faf1c67f1362fbbf4 : SUCCESS in 41m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b5032b2d5cd6473c98ca624f2d39e301 : SUCCESS in 37m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8eb9bed276384776b7475dc2be7083dd : SUCCESS in 40m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/20f1e04e0a4f47ebac58fd403e3e4caa : SUCCESS in 31m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3162b9f77ec1428dbbcf5b985580a25f : SUCCESS in 28m 58s","accounts_in_message":[],"_revision_number":9},{"id":"a6cad8b7cc04066b033aafba107ef1d3a5a0d85e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 17:04:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"9d1451d54e2770f9d8eef69fa2e60dbb07b59ae2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 17:14:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4501239f4789464c8ae7401617366bef : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"20aa42b75be349e9d8097b598253d74fedf11619","revisions":{"187942b98cca32d929b0e35864ea0882e39380a2":{"kind":"REWORK","_number":1,"created":"2021-07-16 15:14:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/1"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:14:31.000000000","tz":540},"subject":"Replace deprecated usage of ABCs in collections","message":"Replace deprecated usage of ABCs in collections\n\nABCs in collections should be imported from collections.abcs and direct\nimport from collections was deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/187942b98cca32d929b0e35864ea0882e39380a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/187942b98cca32d929b0e35864ea0882e39380a2"}]},"branch":"refs/heads/master"},"0efd29ae3d539c037b0c078505a4ae6ac04c64ed":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-16 15:18:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/2"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:18:32.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections was deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0efd29ae3d539c037b0c078505a4ae6ac04c64ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0efd29ae3d539c037b0c078505a4ae6ac04c64ed"}]},"branch":"refs/heads/master"},"4b4e8e3197e9acc2ee0cb78a2a4d74ce8a419eef":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-16 15:21:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/3"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:21:34.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b4e8e3197e9acc2ee0cb78a2a4d74ce8a419eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b4e8e3197e9acc2ee0cb78a2a4d74ce8a419eef"}]},"branch":"refs/heads/master"},"3116b58d36fce46589e38be8c65f82fa673ba92c":{"kind":"REWORK","_number":4,"created":"2021-07-17 03:46:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/4"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-17 03:46:24.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3116b58d36fce46589e38be8c65f82fa673ba92c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3116b58d36fce46589e38be8c65f82fa673ba92c"}]},"branch":"refs/heads/master"},"43c7131f494f2505fa7c8bfaef5e1838aa9f2b94":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-19 00:34:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/5"}}},"commit":{"parents":[{"commit":"a1e878b9e280dda675218e67eeb84c65429b2f64","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a1e878b9e280dda675218e67eeb84c65429b2f64"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:34:20.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43c7131f494f2505fa7c8bfaef5e1838aa9f2b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43c7131f494f2505fa7c8bfaef5e1838aa9f2b94"}]},"branch":"refs/heads/master"},"8cafe682487a113330e112bd34c82108a17620cf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-19 00:39:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/6"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:39:28.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8cafe682487a113330e112bd34c82108a17620cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8cafe682487a113330e112bd34c82108a17620cf"}]},"branch":"refs/heads/master"},"c5a9d421dadc3cf18d71eba477d29f96976f6dc6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-07-19 02:32:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/7"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:32:27.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5a9d421dadc3cf18d71eba477d29f96976f6dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5a9d421dadc3cf18d71eba477d29f96976f6dc6"}]},"branch":"refs/heads/master"},"9db8e956a3941d21ffd488909076a0893b8341a5":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-03-02 09:29:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/8"}}},"commit":{"parents":[{"commit":"55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8","subject":"Use own util method(safe_load) to load yaml files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-02 09:29:55.000000000","tz":0},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9db8e956a3941d21ffd488909076a0893b8341a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9db8e956a3941d21ffd488909076a0893b8341a5"}]},"branch":"refs/heads/master","description":"Rebase"},"20aa42b75be349e9d8097b598253d74fedf11619":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-03-02 09:30:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/801078/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/801078/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/801078/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/801078/9"}}},"commit":{"parents":[{"commit":"e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b","subject":"Replace deprecated arguments of RequestContext","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:13:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-02 09:30:29.000000000","tz":0},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ide8aa0323d9713c1c2ea0abf3b671ca4dab95ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/20aa42b75be349e9d8097b598253d74fedf11619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/20aa42b75be349e9d8097b598253d74fedf11619"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
