)]}'
{"id":"openstack%2Fmistral~801218","triplet_id":"openstack%2Fmistral~master~I8ecedd341ba45d1919c9073dda8b2fc390306a52","project":"openstack/mistral","branch":"master","topic":"bug/1936667","hashtags":[],"change_id":"I8ecedd341ba45d1919c9073dda8b2fc390306a52","subject":"Fix unit test failures caused by type mismatch","status":"MERGED","created":"2021-07-19 00:34:29.000000000","updated":"2021-08-03 15:45:32.000000000","submitted":"2021-08-03 15:44:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801218-bug/1936667","meta_rev_id":"afbc80aa688af3842b9e65ea274c68733bb1ef26","_number":801218,"virtual_id_number":801218,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-03 15:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-07-23 06:22:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-08-03 14:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-21 06:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-08-03 14:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 02:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-21 06:16:14.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-07-23 06:22:07.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-08-03 14:49:41.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"63a8f4107d35ae0db908e5079272c2cd3097b10f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 00:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac6765520820ed9cab316bcd3ce95e1561513a11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 00:39:36.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2f2272f88a61361bd5781f6e08674c02dc6d2eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 02:26:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790ce9dc240c485e949978e61774dd33 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e36f356e3304861981a9bff52fe66ef : FAILURE in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37697a50eaf44a22a54e7e4a7f9e463f : FAILURE in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4ea905e5a214c6fa4b3e3172c0fc2d7 : FAILURE in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a015531936514dfca742da148ba8cfc1 : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db893fa6ca96486e8bf3a566e840082f : FAILURE in 8m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9cc75dda3e424bb3a2a2defd19cae3e5 : SUCCESS in 39m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b16f0c2559484dff9d62c3683ca314bf : SUCCESS in 42m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7748b4313cd4f25abc3930e220a9133 : SUCCESS in 44m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/28eee6c110284610aa5af7065f1a3a98 : SUCCESS in 37m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5608fe889bf34f59a575eaed83cef167 : SUCCESS in 43m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ca006f8a0a494d229d296635bfed98c0 : FAILURE in 32m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cc5a831b347d44828e534348bd51a056 : FAILURE in 28m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d87be299ef614f6888c5024169caeb41 : SUCCESS in 1h 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"293ae012409df5e9f86602efd65736d9f4dba469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 02:30:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"520353d0dcb3bfdf7264d398b612a11b3ee7d36c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 02:32:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"12bd0547a75458c1acc69a37136e119abeeffbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 06:44:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5dd62d8e6584a9b929e5e1768068c83 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf3145d9341a4aefbb79cde26551e891 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13bc89e2362b4801b1703ae5afa24ee0 : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebff050072414bbcb0ea87c1ca105fe7 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167e86ca2e3b48dda62153f09a6bdaf7 : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d28fc439c814160af8c14307ccf9fc7 : SUCCESS in 9m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dd2a9aa8809840dda2ae4fd7c9d41a3e : SUCCESS in 35m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a4b8c2198fd451f9feed3a144e48639 : SUCCESS in 38m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/662443de014949ab8092127691a77444 : SUCCESS in 41m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/55ea24e6924744ef8fb217a82599c3bc : SUCCESS in 37m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c8e81c1cc38c42868fdbd1037f5f0e89 : SUCCESS in 35m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6015e74b5b9b4c62bfc3c83b3bfba7e2 : SUCCESS in 30m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/447e6142bc33433ba61d7aff4e3faec3 : SUCCESS in 28m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1aa8354bc67348d9bf93c89e7b18289d : SUCCESS in 2h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5fc8cf8bce4eea80bdf9e49fe63b33800d8af164","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-21 06:16:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"47061116f56163ff0e0673215c858c102f257288","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-23 06:22:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"70f3502db86893c87656d95415dcdfab9fc0568a","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2021-08-03 14:49:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ebad23d64ce9e82bdbd413d8988223ba0e4177e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 14:50:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"de9e5817051e055212fd5b58cf5f8170f4ceb09a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:43:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b711f01d7f214490a508d26c5e5db5c5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4171ab202a714315bafea05b98810994 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d5348595dbb45aa9d0b6dca3a0626fb : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5d4d64b2ac4f69861e8cc21fb1770f : SUCCESS in 7m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/19057fee565f441985edb7e37bc00f5d : SUCCESS in 47m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b5a164f30d44678a46f982b2f7e6992 : SUCCESS in 45m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb1270a6295d49b1a8490f592aefa317 : SUCCESS in 42m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6d7022f8c0df454c8f6f04cf2a7130f1 : SUCCESS in 45m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/38124f47cfed4726b511c57e7b522a52 : SUCCESS in 45m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/49dde1d3ee414ad4b400908845a63646 : SUCCESS in 29m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e2d94f015cfb4a1286b7d10fb98253e7 : SUCCESS in 28m 25s","accounts_in_message":[],"_revision_number":4},{"id":"a21becdc0ca0da8579593b1378be5a5d482370a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:44:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"afbc80aa688af3842b9e65ea274c68733bb1ef26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:45:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce00d27cfd2c49ebb912e8871dbea1dc : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","revisions":{"a1e878b9e280dda675218e67eeb84c65429b2f64":{"kind":"REWORK","_number":1,"created":"2021-07-19 00:34:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/801218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/801218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/801218/1"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:31:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:34:20.000000000","tz":540},"subject":"Fix unit test failures caused by type mismatch","message":"Fix unit test failures caused by type mismatch\n\nCurrently unit tests are failing because of the following type mismatch\ndetected.\n\ntesttools.matchers._impl.MismatchError: \u0027{\u0027k2\u0027, \u0027k1\u0027}\u0027 is not an\ninstance of list\n\nThis change fixes that error.\n\nChange-Id: I8ecedd341ba45d1919c9073dda8b2fc390306a52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a1e878b9e280dda675218e67eeb84c65429b2f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a1e878b9e280dda675218e67eeb84c65429b2f64"}]},"branch":"refs/heads/master"},"9f74ebc6fa97e08d6f46566f8543429d4927ab5c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-19 00:39:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/801218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/801218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/801218/2"}}},"commit":{"parents":[{"commit":"8cafe682487a113330e112bd34c82108a17620cf","subject":"Replace deprecated import of ABCs from collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8cafe682487a113330e112bd34c82108a17620cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:31:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:39:28.000000000","tz":540},"subject":"Fix unit test failures caused by type mismatch","message":"Fix unit test failures caused by type mismatch\n\nCurrently unit tests are failing because of the following type mismatch\ndetected.\n\ntesttools.matchers._impl.MismatchError: \u0027{\u0027k2\u0027, \u0027k1\u0027}\u0027 is not an\ninstance of list\n\nThis change fixes that error.\n\nChange-Id: I8ecedd341ba45d1919c9073dda8b2fc390306a52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f74ebc6fa97e08d6f46566f8543429d4927ab5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f74ebc6fa97e08d6f46566f8543429d4927ab5c"}]},"branch":"refs/heads/master"},"b1790cb682fcd982f612c03877d2538565d84078":{"kind":"REWORK","_number":3,"created":"2021-07-19 02:30:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/801218/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/801218/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/801218/3"}}},"commit":{"parents":[{"commit":"8cafe682487a113330e112bd34c82108a17620cf","subject":"Replace deprecated import of ABCs from collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8cafe682487a113330e112bd34c82108a17620cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:31:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:30:45.000000000","tz":540},"subject":"Fix unit test failures caused by type mismatch","message":"Fix unit test failures caused by type mismatch\n\nCurrently unit tests are failing because of the following type mismatch\ndetected.\n\ntesttools.matchers._impl.MismatchError: \u0027{\u0027k2\u0027, \u0027k1\u0027}\u0027 is not an\ninstance of list\n\nThis change fixes that error.\n\nChange-Id: I8ecedd341ba45d1919c9073dda8b2fc390306a52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1790cb682fcd982f612c03877d2538565d84078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1790cb682fcd982f612c03877d2538565d84078"}]},"branch":"refs/heads/master"},"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-19 02:32:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/801218/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/801218/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/801218/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/801218/4"}}},"commit":{"parents":[{"commit":"22e822dde29dd68c5868f615629b84b53841c4e6","subject":"Remove references to sys.version_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/22e822dde29dd68c5868f615629b84b53841c4e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 00:31:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:32:27.000000000","tz":540},"subject":"Fix unit test failures caused by type mismatch","message":"Fix unit test failures caused by type mismatch\n\nCurrently unit tests are failing because of the following type mismatch\ndetected.\n\ntesttools.matchers._impl.MismatchError: \u0027{\u0027k2\u0027, \u0027k1\u0027}\u0027 is not an\ninstance of list\n\nThis change fixes that error.\n\nChange-Id: I8ecedd341ba45d1919c9073dda8b2fc390306a52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
