)]}'
{"id":"openstack%2Fmistral~801223","triplet_id":"openstack%2Fmistral~master~I6cca093050c2ef696cf17174efc7b46330c10143","project":"openstack/mistral","branch":"master","topic":"yaml.load","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-03 13:54:36.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I6cca093050c2ef696cf17174efc7b46330c10143","subject":"Use own util method(safe_load) to load yaml files","status":"MERGED","created":"2021-07-19 02:41:02.000000000","updated":"2022-03-03 17:01:54.000000000","submitted":"2022-03-03 16:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801223-yaml.load","meta_rev_id":"5e8c9284f241d8c6e075bfedb340ecc68b001f0c","_number":801223,"virtual_id_number":801223,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 16:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2022-03-03 17:01:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2022-03-02 09:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2022-03-02 09:54:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 04:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-21 06:17:00.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-08-03 14:43:05.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2021-08-04 07:46:44.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"}],"messages":[{"id":"96b1cb3d812c70886ee5432cd907d63e8e2f5394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 02:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e77b156bc2c8d0318cc144d5cf7ca428bade1cf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 02:41:25.000000000","message":"Topic bug/1936667 removed","accounts_in_message":[],"_revision_number":1},{"id":"66deef9cf18e8da080588780998aa896c1d24295","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 02:41:30.000000000","message":"Topic set to yaml.load","accounts_in_message":[],"_revision_number":1},{"id":"adb16f1334c62e0d3fadda54dcc934b4b839997d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 04:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fed6b7e02240f1a75f9a83794b5177 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97a3b69f06424f5fb98a9cbab1edd041 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/440296f028694081a00d2952049f8cd2 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2da9fd983874db985ff5e999505fa6b : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cee86150a8e4794a3fea7a82be8d356 : SUCCESS in 8m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/704262e28d8540bf93eea4ab210bdacb : SUCCESS in 9m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8e873e7f0e7743f0a839b00dbfab9211 : SUCCESS in 35m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a895aae6f6142d3af0bc791ede735e6 : SUCCESS in 36m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d722a4e7cd174c19b62453fa01cbb171 : SUCCESS in 36m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ca6fec68405e4c9d95b7fd173a984de5 : SUCCESS in 35m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/747dddd13cc348299fff940fc43b42d1 : SUCCESS in 37m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/69edb9a47cbe4f31993abd89bdc63ec2 : SUCCESS in 29m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4676862cf47940789b6cbb70345c3b08 : SUCCESS in 27m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1afceb12a93e418cb0d7cdbef63e65b4 : SUCCESS in 1h 48m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e685bc29d2f9e93447ad48201244b8550d8ee117","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-21 06:17:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97218fec2ce1f607eaf980c100b4c092431b6d38","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2021-08-03 14:43:05.000000000","message":"Patch Set 1: Code-Review-1\n\nyou can use mistral.utils.safe_yaml method, like we do in mistral\u0027s code","accounts_in_message":[],"_revision_number":1},{"id":"44caf092538bea0b25f4f1517ab7ff35a3e0fd4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-03 14:52:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dce5dbbe63d8444d113267b33e43187bc8952c70","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-03 14:52:52.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e you can use mistral.utils.safe_yaml method, like we do in mistral\u0027s code\n\nIndeed that is a better choice ! Updated the change accordingly.","accounts_in_message":[],"_revision_number":2},{"id":"c40cedfc0bc45d387d1d47a53358d1b977c97ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 18:45:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007a462cd75d471a80d27fcd049f84d4 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a885393275448e8b5861f1f997aed0 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342b096fc04449fcb6d976e1de67d458 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/732da30c7df8437191f6cea09edf4961 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c1422c5593415b88ed32d3a4a273f1 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/299b013c5fab487ab539725301ea5167 : SUCCESS in 10m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7259714a0966409cb5c8f65a83ec7720 : SUCCESS in 44m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c88c9a82994344429c169b37023ee75e : SUCCESS in 42m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad69e193e7f2441297cdada9614b93ff : SUCCESS in 41m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c2100f49fa5140e58828cd4d01df2502 : SUCCESS in 44m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e1a66379dab4d528337c26bf5c6cc57 : SUCCESS in 44m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9a0b60eeb4b643db8a6c5af687c9a871 : SUCCESS in 29m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fabf35b9f8df4226b9bf00952b7b2989 : SUCCESS in 27m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0ab5cd7722be4b0f8c813d433671b430 : SUCCESS in 1h 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31cece9b3ac6ac37d0753beacba0ecdc5a8a13b0","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2021-08-04 06:53:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"573c9098f809d7cd3c820c772db0b6741c954f08","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 06:54:32.000000000","message":"Patch Set 2: Workflow-1\n\nOops. I think I should update the commit message as well.","accounts_in_message":[],"_revision_number":2},{"id":"8ef6debe239633ef1f880aa11000c66b675bbb4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 06:56:52.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"87628b9b583e46f6095913500c64429cb15f841a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 07:31:18.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ee10eff0bf5bc3ed3c2ec34f2f537b2d2c86631f","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2021-08-04 07:46:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6cddbf3a8c6593d9e4ca46eb9269553959dfc66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 09:37:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ee825173374347b810a314a566f1ad : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e9859351ea842efb9e898e7619ced90 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599ef9969f44476d8f54e08ad26e08a2 : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a348f0ef7984499875cef3645dcd1cb : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f368beff9947618204dfd1e749e184 : SUCCESS in 9m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3abd380a3a0242d0b55935e4b0a98f72 : SUCCESS in 9m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/36d75367faa44d4883e0b8e444173540 : SUCCESS in 39m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a60f6f75bf5450aa95a93e6612c5c84 : SUCCESS in 38m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/732a8a7f9e424530b8353388799d4758 : SUCCESS in 44m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9bf17578936f47128bee152c5f76edb1 : SUCCESS in 46m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/acb7f6fe89f24448b03ed58fa45ba414 : SUCCESS in 37m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/034cfbf7baad4d1aba89223713ded095 : SUCCESS in 29m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/43a5688ddf7f40ec904223470ec80b9a : SUCCESS in 28m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/da7ab12bbfc6484687cee0734d7db856 : TIMED_OUT in 2h 04m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a9a943c7ce5252e6267aafec60e43d83b11cb29a","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2021-08-04 13:02:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5904206ede02df7e66a1e2e2c5b832376ec4e465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:34:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8efcc17cf1e60fa7a9347d75804c44896d98d9d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:28:47.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"7e5080030bcf8d7f8643c2d29967fdef0cf27068","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 09:29:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34cb2f8d35d242232c9c98737a4cebe81d3ec9c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 09:29:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"434ee9b6369abda48b13720a4952f1aa6523f387","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-03-02 09:54:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"745d8b8381403582116e6eb6c88c9d7e5b7fba8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:28:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b405cd3e86cb48788db37fd7f5a89754 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/995a3fbace0c4a5b9e35f459f3891fa3 : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd5605be972c4055895ca10444c2f3ab : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b7de021608748d88f9c3a9da472aba8 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b75719d85c4dab981bc57641345e89 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f7df1a4517464caa06285676317b51 : SUCCESS in 10m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1f032e2f04bc46cbad46bef8e57f432e : SUCCESS in 45m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1682952a5dda40819a1c4c32fc74dfdc : SUCCESS in 51m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46e61e25b4d04788a22e334a95ebd10d : SUCCESS in 45m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fce19997a3a043c6bc00655b7d81b739 : SUCCESS in 49m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b9868e91ac824830a31154941443f905 : SUCCESS in 45m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f3fa1487a3a148038c3f567b808f4cf8 : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/30660612a7194f859b0f2fdf23fe67db : SUCCESS in 28m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c1487436304b45b7946a20b5f017d864 : SUCCESS in 1h 48m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"857867a42d98ea2e1788bdb7e45574d583750c2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:29:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"18097f74b1c99c30761628045dce17ce000af31e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 14:19:41.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f20ee561ad4295a453d10ed9eb77dd : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c87ac17ff48945f19f2fe5c63903963c : SUCCESS in 10m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4854d6908247d5bde89755fd7f4a7b : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8553c7d792a84fa7b476ab128b01b513 : SUCCESS in 9m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/542f3faf44b44f418bbb2a6c9ea3ef9f : SUCCESS in 33m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f395054555f46b49beabe003266c3c8 : SUCCESS in 35m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8a9313503ff43c9a96607865ad2e63b : SUCCESS in 35m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/67455424fd0845b2a9a81efe4dceb290 : SUCCESS in 36m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b7adba2a96054e7db81a7c3effee29b6 : SUCCESS in 37m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/901f2b1d405d4f26894be0fc62ed17ba : FAILURE in 31m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c2d39bc46a194b86be3fd71203eda535 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":5},{"id":"7902400019d7caacf9f801e51e6ab838d671c334","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 14:56:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ef57760159d382e1a8dbe0b3a77f8a1e27430d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:58:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5949b4fd39b84c6a9481c06f72fa2785 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38250d900844e35a99c76e02a50a937 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36656b4c6d4a4c3c844dfa0b86a916e8 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/770de949e0a14329bbf725427b9932ba : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2579b9ac8442869edf34d7f1c9f0bf : SUCCESS in 8m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecb46aaa2adb469484acb571be7c671e : SUCCESS in 11m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ae1892a77b8e41a2be38316d70d250ec : SUCCESS in 39m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1f1f42e59d44005bb49e824b5488d00 : SUCCESS in 39m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04058d31296c4332af7891bca3ff0f68 : SUCCESS in 39m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d4ca4b37e06247528403864e70decfdf : SUCCESS in 38m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2c5ef81d6c33430d8f3c2ae8c98051c8 : SUCCESS in 36m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f39d9934931343e6a0e8227d9424abfe : SUCCESS in 30m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5843dd4fe9c44c468a91a63716acc8ef : SUCCESS in 29m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4b5568a0298d45cc9cd37938fa27ed51 : SUCCESS in 2h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"28cbe793caec7d522dbeb2634697fd138a6f36ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:59:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9048314036b2992156811808121149bb041c2de0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 18:01:49.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f50f8a9d9c46f38601e635309d2bff : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78a3dbc103ef444e8bc39a4a0baaaf4d : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9406bbf9812c46918ec39d6fec0e62cf : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372dba0b67c141cf9d234af21047efda : SUCCESS in 8m 34s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/145ec389f0d64f7ea9da32e07c7b3bb3 : SUCCESS in 53m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1045749aff0845cabfc54bc8dea1bf60 : SUCCESS in 50m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbef4a0d65c94e3dbc6606c313ee61a9 : SUCCESS in 53m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4d0d590f23944939b42c4b1ec6185912 : SUCCESS in 55m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/95857b11c5e241d99d2117b7f22588e6 : SUCCESS in 56m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/699da6777bab40b68c5d1b2f0f77e679 : FAILURE in 31m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c1479c1c0b4a486b96a711ddeae8bde3 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":5},{"id":"51171c4996438aaf524efcb7f47611b5238f0b69","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 23:50:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"efd60ad0a3c6934f97ec40d173d158c3d95e2981","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-03 01:28:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"d3cb1e93e9cd8e5547faf0b8850361e48c6faf11","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-03 01:28:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"a038e96e0a1c6287f85d95b14b8ae746399a2590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 03:19:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8158efcd70584634ad4ac363788dd078 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbfee0c4425647f5a5f630a4bd637261 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd84aa4d0564549afc082b693b4b70b : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce12673aa53c44f787c3a77f11559a18 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d6737fdbd147c8afee54aaee76b216 : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cc2069419dd48cdbd687a7c5e835894 : SUCCESS in 10m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ffedddb40cd542aa8f460640aaa069c9 : SUCCESS in 32m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b928f1cc89d48a18f78a8e3d6c7385f : SUCCESS in 33m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/515eb72cddca4a6caadb1741e2010ae8 : SUCCESS in 34m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c225e60200b5459ca2af0d96a3534b90 : SUCCESS in 33m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/05fcf48fe7b4481fa719c697bd434fe3 : SUCCESS in 33m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0a825c17faa94f6f9fd147da9f0feebc : FAILURE in 29m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/970a4344fbc341f898932c0c0f8d1c2c : SUCCESS in 28m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/24cc4cba68e74b79a3aa4007408aaae8 : SUCCESS in 1h 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"582ec514699c6b268725893d1299cbcd00decf9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-03 03:20:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cc7111a1f6604831b404367297cf1087d2123e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 04:55:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc57d4f02e994b9b86d6efb74f3711fc : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a69edb6cec34c7bbf0e7ace37f66118 : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd1dca90821146e2a5f3465dd8ce14cd : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0c5082830f4426b8ce42037f5e456f6 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a11aaeb0d5e4d4da17575e0ed98c426 : SUCCESS in 9m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1334b83c4fe4c38a06e1d65e32fb8f1 : SUCCESS in 10m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3ae847a1e4534a79b4c160d831c9d9c6 : SUCCESS in 46m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb548d2cd334b58b6d4944bcd4d7101 : SUCCESS in 45m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/131b20df007646e2b14b5266b317f13f : SUCCESS in 46m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d357dbcf4b914e509c2a0512e1d1d745 : SUCCESS in 45m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0f6db16c294b4a2bb5e9cc550b1f513d : SUCCESS in 46m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6642f81334e049b4bb0cc75ebc194272 : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b434cce38855471195a5a8ce3827e5b1 : SUCCESS in 28m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/026f3baf992d484b9739b94c3c2bad97 : SUCCESS in 1h 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c1638bb20f86c0f4207dce5a2ce5efc726a8ca58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 04:56:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"39e2a8798f3c3e705fa31cad2ed24c00c8bacd1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 05:38:28.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc68fee04834f378b6d2b95e0ba8a39 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e28e49a1f0c5401795abd66891855709 : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf95cc952e164a649dd7d330bda59729 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72e3691e29e418cbdc17b740c45432f : SUCCESS in 9m 38s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d48380ea06cf40948af94878445cf8db : SUCCESS in 38m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c289d2deee442fb8e9f2281b3311d83 : FAILURE in 41m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a499eb8986ce4fe7a212444642d1fa16 : SUCCESS in 41m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3177d36dbc2e4099bcb5bc50ac305781 : SUCCESS in 38m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/29faa03679ff438e8d30580475891e04 : SUCCESS in 38m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d1b508568988468ab3195e5de511e6f8 : SUCCESS in 30m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0777b86e84354fc1bed170fb995111f8 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":5},{"id":"713bd6bf3d208ccda738be033037c8a86c3db52c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-03 05:55:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f46ad8d004f10301523cc637b385696aeab157d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 10:17:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e5909103f140cda98507de42e7f465 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c235aab498df49c8b4d75a7d65d26e71 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93048ec786dc454fbee7b7bc8c6d7668 : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e97817ba73ec4d5292e7cf8a9e2f01f9 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4dd8eb27e4f46259c8f428e657b51d7 : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f843733059ea45ffbbee84e91c95b628 : SUCCESS in 10m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/65b295c097134ab788b800559c72fcf8 : SUCCESS in 35m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78ca9126d8794f44b17cb82fa9fa1cd4 : SUCCESS in 33m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69cbe4b0710c46c59dd4da32b8c9ef4d : SUCCESS in 35m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/556424a8c6c747579d5258275b455ee2 : SUCCESS in 33m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9cdc7a6c5e13455d8225cbf55d32e75c : SUCCESS in 47m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bb17969a83404fae8b7ef29febdfcd7f : SUCCESS in 30m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/749952e4255f4a019123025162ac909d : SUCCESS in 28m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f0b349b978874f4f8155fc4749808c88 : TIMED_OUT in 2h 05m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"66116c364f647378f2bf91a6352eefa364cb0ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 10:18:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"145dd5c0d4db3f7e74a8bc2ff574742ce20a99e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:18:22.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51fc92e3a614b6a8411c30707c27cc2 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c6fd6b30dd742d8b7f8bfc9b63791f1 : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28dc83caeac6433589efb37ccd600805 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9199b9639544bca2457a7cb3c6488d : SUCCESS in 8m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bd8be3b571b94121a68206f5f55c3882 : SUCCESS in 54m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f23771e87d694ec48da20ff55303237a : SUCCESS in 55m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/defae360edcd4bb8bb5f566f9ad3a3cc : SUCCESS in 45m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bb5c414cc44f451ca2c79c901dc0eda9 : SUCCESS in 33m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8d20281560144caf84c4114f27b7767f : SUCCESS in 51m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cbca273195274ae683c705cce2fe5f8e : FAILURE in 29m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/76e97c7dd7b4437ebf01f9afc68b21e0 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":5},{"id":"40a830139d3a43aa34b0e4082bc8f021e247c5d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-03 11:47:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"964a38adce860a362ce1c3272bd2a8f524ce24a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:40:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e0ecf4271b478fa4dfbf37156cd298 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/505558cec606489fbb68ed4dd564b789 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dad2cb930124d24b19361ec6f83984a : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd229c4c94c4e63a9b7f02ff963094f : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eced09cb8fa34233aee8cc19b9bb7b7c : SUCCESS in 9m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8c80c123f974f8a8ed3ce85da31b1e1 : SUCCESS in 10m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f623889cdcb84b25841044b2d5e5c3a4 : SUCCESS in 44m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5aeff056126041299ea41ae45a4c71a6 : SUCCESS in 40m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0490defd6b649fba616296cab122e19 : SUCCESS in 41m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e52b2ea31bb440c9af826cb2cb9db4d3 : SUCCESS in 40m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9bfcfc42ede14455b68aeff9cdf30be8 : FAILURE in 36m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/190a737656ff4c59a4d489f378c78c21 : SUCCESS in 31m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6673ba7ff6734e68b607e7bceb50b868 : SUCCESS in 31m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1026a23184154b17b391922e8e697330 : SUCCESS in 1h 50m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"294366dad4fbf7b1df783cb509f99187b8b92692","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-03 13:54:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3a993d2a4afe08f92eb2b36dcc892d10426e5372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:56:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4af228b25344846864871151ff8e434 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7cf07276c3c468bafefb59be0c50141 : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/919bbaee212341c3aa52505badb3239b : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8272c7d9dcfd4ebeb9eeb80e83dcc3d3 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9354eb1a8802477db75cc043fa0be027 : SUCCESS in 8m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ca0d5cef74142d6a7d8a01511ef5ec5 : SUCCESS in 11m 20s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a6e1d83970b34c668313fea42b9910de : SUCCESS in 47m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c464a2230a72457bb4218a83cbbd1ce4 : SUCCESS in 42m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ea6a8da9eab4c6c86bc41c466a79ed2 : SUCCESS in 38m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/db4402a8ee6641b29e0416a45b640a12 : SUCCESS in 39m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ed9c805ea0c841ffa1304df07bd3e426 : SUCCESS in 45m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e11686b678ff488ab91ffcff78ec1537 : SUCCESS in 31m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e043d5098b494f17986e1758309d9b61 : SUCCESS in 29m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb2a28eb32e447c5b9363e17dcf6d3b7 : SUCCESS in 1h 57m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a777441bad2ff559b03a1191130abee591f70ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 16:00:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7fde9558d8ffb5693c4daffbf87b96acf66356cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 16:59:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6030bdb79e4dfe985cd554412b8e54 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/387824cbf18943ee83f0654a4e858834 : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6ef634e0f7c43229bd3a4889b578655 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2594354657b04be397c6dffdcec7e627 : SUCCESS in 9m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0397b1766b514b92beaf4f268efa6480 : SUCCESS in 41m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec48be7cdb814678bdc5d5b67ac71c57 : SUCCESS in 43m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abbacdd310b04994b8c61bee7609fe63 : SUCCESS in 54m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dab8e8a778f34a24acae9189522a6ddf : SUCCESS in 48m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e1402ac6987448bdbcc546a752cae3da : SUCCESS in 47m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/263195d2a9c44ff8b016d454bd62d3a1 : SUCCESS in 31m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b87933a6bd5d46be8907206d70e67a0e : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":5},{"id":"6a4e9cf5c84fb937401fa98b398821cac47541fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 16:59:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"eb9a4dd54880b5b863430d5c53b79860da64c701","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 17:01:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13db509326e54c52ab77e1f3c19de2e6 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8","revisions":{"e8fbde4bd38d7f8df54bb9d4e757783367793ac8":{"kind":"REWORK","_number":1,"created":"2021-07-19 02:41:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/801223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/801223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/801223/1"}}},"commit":{"parents":[{"commit":"c5a9d421dadc3cf18d71eba477d29f96976f6dc6","subject":"Replace deprecated import of ABCs from collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5a9d421dadc3cf18d71eba477d29f96976f6dc6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:38:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:40:25.000000000","tz":540},"subject":"yaml.load: Set Loader to avoid deprecation warning","message":"yaml.load: Set Loader to avoid deprecation warning\n\nSince PyYAML 5.1, yaml.load without specifying the Loader option is\ndeprecated and shows the following warning.\n\nYAMLLoadWarning: calling yaml.load() without Loader\u003d... is deprecated,\nas the default Loader is unsafe.\nPlease read https://msg.pyyaml.org/load for full details.\n\nThis change adds the Loader option to get rid of that warning message.\n\nChange-Id: I6cca093050c2ef696cf17174efc7b46330c10143\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fbde4bd38d7f8df54bb9d4e757783367793ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fbde4bd38d7f8df54bb9d4e757783367793ac8"}]},"branch":"refs/heads/master"},"b9b68fcd90729f2ee08d26d19041a266e0adbae0":{"kind":"REWORK","_number":2,"created":"2021-08-03 14:52:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/801223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/801223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/801223/2"}}},"commit":{"parents":[{"commit":"c5a9d421dadc3cf18d71eba477d29f96976f6dc6","subject":"Replace deprecated import of ABCs from collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5a9d421dadc3cf18d71eba477d29f96976f6dc6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:38:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-03 14:51:20.000000000","tz":540},"subject":"yaml.load: Set Loader to avoid deprecation warning","message":"yaml.load: Set Loader to avoid deprecation warning\n\nSince PyYAML 5.1, yaml.load without specifying the Loader option is\ndeprecated and shows the following warning.\n\nYAMLLoadWarning: calling yaml.load() without Loader\u003d... is deprecated,\nas the default Loader is unsafe.\nPlease read https://msg.pyyaml.org/load for full details.\n\nThis change adds the Loader option to get rid of that warning message.\n\nChange-Id: I6cca093050c2ef696cf17174efc7b46330c10143\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9b68fcd90729f2ee08d26d19041a266e0adbae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9b68fcd90729f2ee08d26d19041a266e0adbae0"}]},"branch":"refs/heads/master"},"71cb419ae6ad382c66fcdd4ff42b61a5eaf2611d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-04 06:56:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/801223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/801223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/801223/3"}}},"commit":{"parents":[{"commit":"c5a9d421dadc3cf18d71eba477d29f96976f6dc6","subject":"Replace deprecated import of ABCs from collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5a9d421dadc3cf18d71eba477d29f96976f6dc6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:38:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 06:54:50.000000000","tz":540},"subject":"Use safe_lood to load yaml files","message":"Use safe_lood to load yaml files\n\nSince PyYAML 5.1, yaml.load without specifying the Loader option is\ndeprecated and shows the following warning.\n\nYAMLLoadWarning: calling yaml.load() without Loader\u003d... is deprecated,\nas the default Loader is unsafe.\nPlease read https://msg.pyyaml.org/load for full details.\n\nThis change replaces yaml.load by Mistral\u0027s own safe_load method which\nis widely used to load yaml files.\n\nChange-Id: I6cca093050c2ef696cf17174efc7b46330c10143\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/71cb419ae6ad382c66fcdd4ff42b61a5eaf2611d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/71cb419ae6ad382c66fcdd4ff42b61a5eaf2611d"}]},"branch":"refs/heads/master"},"fc29a461521e1f15740f7bf81fccfb7edd4d95e6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-08-04 07:31:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/801223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/801223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/801223/4"}}},"commit":{"parents":[{"commit":"c5a9d421dadc3cf18d71eba477d29f96976f6dc6","subject":"Replace deprecated import of ABCs from collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5a9d421dadc3cf18d71eba477d29f96976f6dc6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:38:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 07:31:12.000000000","tz":0},"subject":"Use own util method(safe_load) to load yaml files","message":"Use own util method(safe_load) to load yaml files\n\nSince PyYAML 5.1, yaml.load without specifying the Loader option is\ndeprecated and shows the following warning.\n\nYAMLLoadWarning: calling yaml.load() without Loader\u003d... is deprecated,\nas the default Loader is unsafe.\nPlease read https://msg.pyyaml.org/load for full details.\n\nThis change replaces yaml.load by Mistral\u0027s own safe_load method which\nis widely used to load yaml files.\n\nChange-Id: I6cca093050c2ef696cf17174efc7b46330c10143\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fc29a461521e1f15740f7bf81fccfb7edd4d95e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fc29a461521e1f15740f7bf81fccfb7edd4d95e6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-03-02 09:29:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/801223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/801223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/801223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/801223/5"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 02:38:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-02 09:29:15.000000000","tz":0},"subject":"Use own util method(safe_load) to load yaml files","message":"Use own util method(safe_load) to load yaml files\n\nSince PyYAML 5.1, yaml.load without specifying the Loader option is\ndeprecated and shows the following warning.\n\nYAMLLoadWarning: calling yaml.load() without Loader\u003d... is deprecated,\nas the default Loader is unsafe.\nPlease read https://msg.pyyaml.org/load for full details.\n\nThis change replaces yaml.load by Mistral\u0027s own safe_load method which\nis widely used to load yaml files.\n\nChange-Id: I6cca093050c2ef696cf17174efc7b46330c10143\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
