)]}'
{"id":"openstack%2Fmistral~801853","triplet_id":"openstack%2Fmistral~master~I24a8963d112a24674ad5699685118d019ef1ed56","project":"openstack/mistral","branch":"master","topic":"context-attr","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2022-03-04 10:00:50.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-03-02 08:07:59.000000000","reason":"Adriano Petrich replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-04 10:00:50.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-03-02 08:32:57.000000000","reason":"Eyal replied on the change"}},"hashtags":[],"change_id":"I24a8963d112a24674ad5699685118d019ef1ed56","subject":"Replace deprecated arguments of RequestContext","status":"MERGED","created":"2021-07-22 13:30:48.000000000","updated":"2022-03-04 10:02:15.000000000","submitted":"2022-03-04 10:00:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801853-context-attr","meta_rev_id":"d33aaec88330a2a0bca8960dc637d0eda49ce4aa","_number":801853,"virtual_id_number":801853,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-04 10:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":2,"date":"2022-03-04 08:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-04 10:02:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":1,"date":"2022-03-04 08:48:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-22 17:19:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 00:22:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-03-02 00:22:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-03-02 08:27:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"522e8c65093094e06f21a1bd1d23ae51487dee64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 13:30:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d16942380f73540b3296d8aa994b61eff26c9e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 13:32:55.000000000","message":"Topic oslo.context removed","accounts_in_message":[],"_revision_number":1},{"id":"4092390b4949824d44e6e222b0769144974b899c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 13:33:00.000000000","message":"Topic set to context-attr","accounts_in_message":[],"_revision_number":1},{"id":"8ce4806c9d4c58a28ce99b943b17ba570a43929f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 13:37:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e210a3586f564409e2870368f385a8c38d3279c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 13:42:21.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dffc3e0d7d7a2289956426c093f9390e61c6d3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 15:25:34.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"11b261146c7faf793028f9c5a4d5b9ac9411b464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 17:19:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5382ba9ca84a65b093ee78b9c8052b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7e3225246d8449698d084525039a25c : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aacf15de8199443fa114b922764522d3 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a7a5af19abf4751b9960a066d7385f1 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd7581490704c088e83356c5ba1037d : SUCCESS in 8m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f16c362164c46e5a791e5bbf68b979a : SUCCESS in 9m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7b69cdaafdd7494a80c48fa14e60445a : SUCCESS in 41m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c87f149627945bb9558cc5d9a7d98ef : FAILURE in 19m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a626eb341cce427ba1be3ee34b8ef168 : SUCCESS in 46m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/585ebf058e244d0a83b61bc6c235c819 : SUCCESS in 42m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c8a06ed68c5f4d49b55614f13e0ae59a : SUCCESS in 42m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f5928c1e921546ae9305cddd12f54986 : SUCCESS in 29m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a63a92f35c8b403f89867d5e50e72da3 : SUCCESS in 27m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0a864bb7704141ad803baede6a32b33a : SUCCESS in 1h 52m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e18897d471c14a3e05fdccab4a294912dfdb1ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:27:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"545130fe26f02cd5a505eaf6020c8b336920b87a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 19:23:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71990e57a4b432a890ade2738f0a71a : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d1cdc2324974aa893459717b8141dab : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c124e1b65547018b0ae666d9393647 : SUCCESS in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a42dab555e5496bb8fcd56400b92144 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a571e250a8f492883c75c07adaf878a : SUCCESS in 8m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebda90ff40694f69ad93bc87c92d8b27 : SUCCESS in 9m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/27f8ace643fc4f18bd5beb43261b1072 : SUCCESS in 49m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dffc76b02bed41bb979b619cca77a9c3 : SUCCESS in 45m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5f99e4151614f3bbaafffad9a66e43f : SUCCESS in 44m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/13d6bad9f7d141438510971f31fe625c : SUCCESS in 44m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/38a6d849b95f410fb80588f4d7be54b2 : SUCCESS in 44m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9437a3dd95734678af9198f63bf0e8be : SUCCESS in 28m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/53cb8c7350f24d349ee1b666a1d13dca : SUCCESS in 27m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb7e6adc25c84ef3ad5bba814ce69a23 : SUCCESS in 1h 54m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0ed62a79ee0ea1608dc2c58a2069416c9babfc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-03 14:52:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2f16352205fcfebd3d7edc38783deb892292dfcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 16:49:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd3683de8194abf956f2e081dedcb73 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4954313dbe6046eba22bc711b38c2096 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599ed412674d4eda89c7875e3a16082f : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87d80a9794a14978b3229488a44a2e64 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8bf04079c24f39b13aa5e23d6dc0c0 : SUCCESS in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85136b3ecb5e40489cf53c1765212b03 : SUCCESS in 10m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ef02fa617873417faaa31a1398ef41f5 : SUCCESS in 38m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4931e4ca92e04fdba862339f69dfb57a : SUCCESS in 46m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b796b706352a438cbd261d8c1010ded4 : SUCCESS in 42m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/16cf63af85754950a9163abb48d1e692 : SUCCESS in 38m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d1a8d0be31bc409891a790c835d7cc35 : SUCCESS in 37m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/35de6052b30547d393700e2293b45c8c : SUCCESS in 30m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4af57b2884f34a11819a99bc74951dd8 : SUCCESS in 28m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9bc98e4e1b214280b680c3bb5e8be474 : SUCCESS in 1h 55m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c9dfe0380ec03a9be1aa01928a3a96f1248c2db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 06:57:14.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1afa365d97bfc3b9c85f09cdab4c8cb953c7289e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 07:31:37.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"7fa462de2cb52c1df05feb22f615f2624499de6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 09:26:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0fd63dda73b4725971d53702b99aa90 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fb3734110a4434f80259ea752cf1a07 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d6eb9d6f0144db6a63a1182a0fa64ff : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c41f99ec1d54247b06aff50d87eda4c : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df0d5973b9e498580bac89efb19b911 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/591c9bdb32614a88908e75a646a4e5f0 : SUCCESS in 10m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ad5405ccb29e4e0c9b7697eb80062d10 : SUCCESS in 35m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d163a199d0c455db01a6c94bd1f6012 : SUCCESS in 35m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c610d42713774a339f290fd538022ee8 : SUCCESS in 42m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bae42abc09344d35b8f6a5c3a70ef079 : SUCCESS in 36m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/02c22f435f2f41fd9962b05dbdf01733 : SUCCESS in 34m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cce91cc65e73496c913e72c5c84896c3 : SUCCESS in 28m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b61f1373cde1469d8e45a9e580458359 : SUCCESS in 28m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/56b54fa5581f4292b857f2a8758fdc4a : SUCCESS in 1h 53m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"248ad84f6470de60dd45aca83c32d11fb53791bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-28 23:23:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cb5ccee5cabee5c456141b71cfc8ebad4f03874b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 01:07:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5edf96e5080949e086aef17ce43bb35e : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e86f5eaf1b1d486b8093af632890317d : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9fc4614cef478b8e5fa043d336e8a5 : SUCCESS in 10m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c5237ee138d424f9e9ae3e39e2005f1 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76add36526944d84891f8ba509fa107e : SUCCESS in 9m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2953c99c9564e73b0b9e066f6dd98cf : SUCCESS in 10m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/12b019627b2741b598bb48c1be77bfab : SUCCESS in 36m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70e8c17e20be4dc08e92e842b818b378 : SUCCESS in 38m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e39b4a2f0c947788dc4ae14c76eb811 : SUCCESS in 32m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a74519055f94427484cc5efea656540d : SUCCESS in 36m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1322e01e4bd74ccf8f01a8945d160dcf : SUCCESS in 37m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ba7d92f093854e5293f3cd6799fe1736 : SUCCESS in 30m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ccbd36338ccb4bf4a322e43eed61dc69 : SUCCESS in 29m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/54dc4d684e774a9fb97e016ac52a1ba2 : SUCCESS in 1h 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2487d13c827059e500a9145f6bccd01ccef80d26","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-03-02 08:07:59.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"4c4cee5dc31e0fc95da1cacce49af936d3191194","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-02 08:32:57.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8916135958cea387d6e1cb08d933f9e1db2d9284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:34:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7e4eb8d7118efc7a6357c85e69a7990b1149f70c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:28:48.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"6764e5b33767d953756f936dbeae2fd7abd55ee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 09:29:24.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"d6a25cd92ea64985e4a1761556907f3de3b44427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:49:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255ebaf1f9cf419ba748531f7625c9b1 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79c8e41603e54f9699051101e5fef6ab : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/778d23d5713c4033b819a0a417f5ffac : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61a3c191cfd14376b549969582d086eb : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38bb37a43e74c76bd8502b3fd107c18 : SUCCESS in 9m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9e2ebd137274ee8a6db9b8135827e82 : SUCCESS in 10m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/76a17a94cf8146beb5f24ad90ad2b9a9 : FAILURE in 2h 22m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0bf850a512848a595e6e9d611ee5000 : SUCCESS in 1h 00m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a32c10b574654b07ab99757df26f50d5 : SUCCESS in 48m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2719e14018ff4bf9b33435d36ac6ed6e : SUCCESS in 1h 02m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0d6de1b1a4a74ab08f93a310290a3f19 : SUCCESS in 32m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/673060c62e4a4b0d867ca4b660dcdc21 : FAILURE in 33m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0f0aca06046348d4a33a9f896b38c943 : FAILURE in 29m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d151631833e84f19bdbf1d013423f07a : TIMED_OUT in 2h 05m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"288da04ead4a35aa2b65acf5e9895f56bcccf844","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 13:56:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ac361d656ddc52b951735ca0022f1451c99a139d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:10:33.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0042fa5a1d54d9891d60dfb1c31a9a6 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b4d519e7624413a2d1ee4d35cea947 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59c71330a9dc48e69e4df25c8b2a877b : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/537e699a75c44ed59f74d521b6d397c9 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2747f5d29a774a37846578e3b3f463c6 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45c878c97ee344eeb98b88cc5968f591 : SUCCESS in 11m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/86da3ed21a014211a02e200e6b9ab7d8 : SUCCESS in 42m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f674021ff5f483ba185297a8c7cc7dd : SUCCESS in 42m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d87558b8a3244c21aa3d56d961a88ca3 : SUCCESS in 49m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/21db2cfc40d14e219cd053edeac89b8b : SUCCESS in 51m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d1827a35eed14b5c975bc061f9260bfc : SUCCESS in 46m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/20c88c1ddc1e44e9b2e7f2ec4f654e06 : FAILURE in 31m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d477a57efd954035aae1de327d3b7f4c : SUCCESS in 29m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/435e8e2f8014400fbda97cd7dcf6fcca : TIMED_OUT in 2h 07m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"25b0ba9c793a76465cc2c7268e617f291f0e91f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 17:30:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9c61098387a6bb992010f0e3bcfb72ae6675fc7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 19:41:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9166f2342a50456cb9b74857d59f3627 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd7fb047dfc242efb77683cd130bd36f : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8bb1d3348ab445fa9e68e3626ecb7c8 : SUCCESS in 26m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dd4f939ed9940a68f9c7eec7ac89dff : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c6a572ab334b2aa1edda3ed31b0ea0 : SUCCESS in 8m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a591e6eb5c5048bb9f2b4729cc2e8ac4 : SUCCESS in 10m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/47fad56fe2ff45fc8af8914d73740d63 : SUCCESS in 47m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6086da12ce2401786db7a4733f8ece0 : SUCCESS in 51m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d61937fe2fbb4556a8ae4401d533e694 : SUCCESS in 45m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/721d98642e0a46a7b73a060119e34018 : SUCCESS in 34m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3e282b4ef1984afe9768a1a76bac9ab5 : SUCCESS in 52m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e0bd6426a44d42d9a7172f1b0fb99746 : SUCCESS in 31m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fa13b21301194f47bf3a09fa9daf1768 : SUCCESS in 29m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e1ae759d32564810b93f1a7843505c0a : TIMED_OUT in 2h 05m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"49763b9cc99a8c48437998f25b34fb95a255c2b3","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-03-04 08:48:31.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"6cca914461c241fba4dc83b2c1b02e50ef2f8052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 08:49:22.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"909b28b2d76ac4353c9019dddf35524468d5883b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:00:50.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88b422ea0894a588536f73c0be1705f : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3905e5a4648447adb899a884db30b475 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b03719f08b241238c22585ad11c501b : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e2ebfb51234f52a30464e1617f9c2c : SUCCESS in 9m 32s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3a0b476e59c94458a9eb6212df2b6632 : SUCCESS in 1h 06m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adade08c243440ee8cec2cb649e6e551 : SUCCESS in 56m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4696a719b51f435eb2c7a83f83c54200 : SUCCESS in 1h 06m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/872f6770e8e04bbcb2cb7bcea5c60675 : SUCCESS in 48m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/293007877eb9497d908aa8cb9927f506 : SUCCESS in 46m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f5a2c68111f54bef9c68d6643b30a9fb : SUCCESS in 30m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e9efae4ddade4301885ec639faddb942 : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":8},{"id":"bc777797b2a2c1f81b6beec8016c04d1f33fa1e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:00:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"5e21fdaa5ef7555dee878de2a629c13e5e7d842d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:02:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c44362c9674d2e9648f6a9644d4626 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b","revisions":{"68f81a984bd84f88251d065856c5e4f972115f69":{"kind":"REWORK","_number":1,"created":"2021-07-22 13:30:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/1"}}},"commit":{"parents":[{"commit":"e8fbde4bd38d7f8df54bb9d4e757783367793ac8","subject":"yaml.load: Set Loader to avoid deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fbde4bd38d7f8df54bb9d4e757783367793ac8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:30:35.000000000","tz":540},"subject":"Replace deprecated arguments of oslo_context.RequestContext","message":"Replace deprecated arguments of oslo_context.RequestContext\n\nThe following two arguments of oslo_context.RequestContext are\ndeprecated since oslo.context 2.18.\n user  : renamed to user_id\n tenant: renamed to project_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68f81a984bd84f88251d065856c5e4f972115f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68f81a984bd84f88251d065856c5e4f972115f69"}]},"branch":"refs/heads/master"},"49a2ff86d41dfb09352ac74855a95687a010b88d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-22 13:37:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/2"}}},"commit":{"parents":[{"commit":"e8fbde4bd38d7f8df54bb9d4e757783367793ac8","subject":"yaml.load: Set Loader to avoid deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fbde4bd38d7f8df54bb9d4e757783367793ac8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:37:25.000000000","tz":540},"subject":"Replace deprecated arguments of oslo_context.RequestContext","message":"Replace deprecated arguments of oslo_context.RequestContext\n\nThe following two arguments of oslo_context.RequestContext are\ndeprecated since oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49a2ff86d41dfb09352ac74855a95687a010b88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49a2ff86d41dfb09352ac74855a95687a010b88d"}]},"branch":"refs/heads/master"},"e58ef3f15cc1034040bc18c3d1ac3572b339be95":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-22 13:42:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/3"}}},"commit":{"parents":[{"commit":"e8fbde4bd38d7f8df54bb9d4e757783367793ac8","subject":"yaml.load: Set Loader to avoid deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fbde4bd38d7f8df54bb9d4e757783367793ac8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:41:55.000000000","tz":540},"subject":"Replace deprecated arguments of RequestContext","message":"Replace deprecated arguments of RequestContext\n\nThe following two arguments of the RequestContext class are deprecated\nsince oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e58ef3f15cc1034040bc18c3d1ac3572b339be95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e58ef3f15cc1034040bc18c3d1ac3572b339be95"}]},"branch":"refs/heads/master"},"df2e55a155e6da1c551fcebf10e8bd5444c52abc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-07-22 15:25:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/4"}}},"commit":{"parents":[{"commit":"e8fbde4bd38d7f8df54bb9d4e757783367793ac8","subject":"yaml.load: Set Loader to avoid deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fbde4bd38d7f8df54bb9d4e757783367793ac8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 15:25:19.000000000","tz":540},"subject":"Replace deprecated arguments of RequestContext","message":"Replace deprecated arguments of RequestContext\n\nThe following arguments of the RequestContext class are deprecated\nsince oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df2e55a155e6da1c551fcebf10e8bd5444c52abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df2e55a155e6da1c551fcebf10e8bd5444c52abc"}]},"branch":"refs/heads/master"},"fcf8c6700467d042a6ad9d94141e657607cb88d5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-03 14:52:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/5"}}},"commit":{"parents":[{"commit":"b9b68fcd90729f2ee08d26d19041a266e0adbae0","subject":"yaml.load: Set Loader to avoid deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9b68fcd90729f2ee08d26d19041a266e0adbae0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-03 14:52:59.000000000","tz":0},"subject":"Replace deprecated arguments of RequestContext","message":"Replace deprecated arguments of RequestContext\n\nThe following arguments of the RequestContext class are deprecated\nsince oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcf8c6700467d042a6ad9d94141e657607cb88d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcf8c6700467d042a6ad9d94141e657607cb88d5"}]},"branch":"refs/heads/master","description":"Rebase"},"d75d0b42153a1313835a5eb2020fdf3ff4f9a8a8":{"kind":"NO_CHANGE","_number":6,"created":"2021-08-04 06:57:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/6"}}},"commit":{"parents":[{"commit":"71cb419ae6ad382c66fcdd4ff42b61a5eaf2611d","subject":"Use safe_lood to load yaml files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/71cb419ae6ad382c66fcdd4ff42b61a5eaf2611d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 06:57:14.000000000","tz":0},"subject":"Replace deprecated arguments of RequestContext","message":"Replace deprecated arguments of RequestContext\n\nThe following arguments of the RequestContext class are deprecated\nsince oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d75d0b42153a1313835a5eb2020fdf3ff4f9a8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d75d0b42153a1313835a5eb2020fdf3ff4f9a8a8"}]},"branch":"refs/heads/master","description":"Rebase"},"154c8dd1e882e812a510ef83f920624d944cba6a":{"kind":"NO_CHANGE","_number":7,"created":"2021-08-04 07:31:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/7"}}},"commit":{"parents":[{"commit":"fc29a461521e1f15740f7bf81fccfb7edd4d95e6","subject":"Use own util method(safe_load) to load yaml files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fc29a461521e1f15740f7bf81fccfb7edd4d95e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 07:31:37.000000000","tz":0},"subject":"Replace deprecated arguments of RequestContext","message":"Replace deprecated arguments of RequestContext\n\nThe following arguments of the RequestContext class are deprecated\nsince oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/154c8dd1e882e812a510ef83f920624d944cba6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/154c8dd1e882e812a510ef83f920624d944cba6a"}]},"branch":"refs/heads/master","description":"Rebase"},"e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-03-02 09:29:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/801853/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/801853/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/801853/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/801853/8"}}},"commit":{"parents":[{"commit":"55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8","subject":"Use own util method(safe_load) to load yaml files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55bb2564ac757f92b4d8f3169d6f1b2f95fef1c8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 13:24:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-02 09:29:24.000000000","tz":0},"subject":"Replace deprecated arguments of RequestContext","message":"Replace deprecated arguments of RequestContext\n\nThe following arguments of the RequestContext class are deprecated\nsince oslo.context 2.18[1].\n user          : renamed to user_id\n tenant        : renamed to project_id\n domain        : renamed to domain_id\n user_domain   : renamed to user_domain_id\n project_domain: renamed to project_domain_id\n\nThis change updates arguments used accordingly.\n\n[1] d78cf592e1e3e7aa0fc99bfdd655e82f5c44dfe3\n\nChange-Id: I24a8963d112a24674ad5699685118d019ef1ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
