)]}'
{"id":"openstack%2Fmistral~807461","triplet_id":"openstack%2Fmistral~master~I17f56606eb43fba46281e9bdc08ace8727455a40","project":"openstack/mistral","branch":"master","topic":"bug/1938103","attention_set":{},"removed_from_attention_set":{"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-08-16 16:36:10.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-16 13:26:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I17f56606eb43fba46281e9bdc08ace8727455a40","subject":"Replace deprecated assertDictContainsSubset","status":"MERGED","created":"2021-09-04 16:13:12.000000000","updated":"2022-08-16 16:37:27.000000000","submitted":"2022-08-16 16:36:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807461-bug/1938103","meta_rev_id":"caa76f19ffc8a48be672ba1cf5593c9996504219","_number":807461,"virtual_id_number":807461,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 16:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-12 14:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":1,"date":"2022-08-14 00:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-12 14:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 18:07:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-05 05:10:08.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-11-30 12:28:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-08-14 00:57:46.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"388971c276d1e8d19f0f2102f96cab42e78bf9d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 16:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0654b0eaaee018dd9c14d69a213d35f0acd2246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 18:07:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7fcaecd0774ca3a70c36568ff74d85 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d6b4d6bb7644c5b95d225b091c59da1 : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d287970bf066441da8ad36dc8d0fc3ae : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60e2d40b360b4d8597ce02b5c0948406 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7281c74c96314e1382bb45aff54c5bb9 : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d104637f5bf742d68104b7363689e47b : SUCCESS in 9m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/822e93142fc545e3a8f9cf02b7ae0122 : SUCCESS in 34m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efeab07787874e118ced4848176263b0 : SUCCESS in 50m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d899a24dd3f84c5189cd7f980b3b54ca : SUCCESS in 42m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/326fa07dd6c44f99966710620501e2cc : SUCCESS in 35m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fee844da6e984241800e419f6a3bbfca : SUCCESS in 44m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c83db572dbf2428fb5bb214505f26de6 : SUCCESS in 30m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9a88bbb629e146d2ad9bc6bbd610bcec : SUCCESS in 27m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/694e8c39acec4a4682b7396ab72ba027 : SUCCESS in 1h 51m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bfe641229be5122c2316d7a152251a56f927e5b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-09-05 05:10:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e90e5cbbf9e291eff984c3593d5a9bea811b724","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 09:41:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1f17c59df5a750f448691fbf53816d77864cc27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 09:42:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"977bcb391530e977d7009e4fec27c3ec3758852c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 09:48:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"da2295a4281ab8d24b27faf506f4102eed6eabb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 09:58:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"08be64a2faa0898a1895e01185a43ab4d6770d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 11:47:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7fa0288e0b04dcb9b1d3c67bbf40d80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a89dac81e4a4907a91d35c7922cc773 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8833e582645e483095fd72d493f32f08 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29369a0569384141a553bd29ed033229 : SUCCESS in 8m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/969c4181ef0b43a89d531950826b55f9 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d464146248ac471f9c697d9c69ce04f3 : SUCCESS in 6m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/04866541de274d56ae8bfc8ea2dcaec2 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88e36b89645e4143b7861f23bbfe8666 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d653c51f2c2347db8218fe0ce3ed5892 : SUCCESS in 9m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6e1abc2adb744d16993eec4e160a06ce : SUCCESS in 39m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7698570205d47159ba07af0f39e0d3a : SUCCESS in 39m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c949ed7a41f743d7a1111b0672261280 : SUCCESS in 37m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c294416cfade4606bd5b49eda594a849 : SUCCESS in 39m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7b2950a466f2429cbd9eaeeea17d202a : SUCCESS in 38m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1bfcddd9223b4077ab3c654ac3b09ae9 : SUCCESS in 32m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/eb4749d338334e10812c8be34ade19fb : SUCCESS in 29m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/05911d2d197d4c639b5bd23bf79a36c4 : FAILURE in 1h 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c9bd673766c071eeafb25ff63d34aaa1d6a7442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-12 11:59:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"79d3672d203cfccc01b6538b948f0d5560c374e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 13:57:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/528ed096ed2d4fcba0384e3b274c2c9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ab174e17274313a83c4f78a4bb71d2 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06f62c8bf8334942a0e9481639ff5d5e : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14a86634995b4271abc1f8d0d5104cab : SUCCESS in 10m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e950c6d3e7134d3e9dc5a18f7aeb6e48 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73100807382d4e15b93c2d4cfc2c19d3 : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33b4710e8f4746ccbb0d8705a5c954fd : SUCCESS in 11m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/220d861f48284ed28f0a0341e243ef4f : SUCCESS in 50m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b7966d07a5a4f6c94eccdf860b92e02 : SUCCESS in 51m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b64e1940968418d9a32d042c791fc84 : SUCCESS in 52m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0c45e97edfc943ea9f4477d8188ff234 : SUCCESS in 52m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b9883349ab0e4a0f93aefd083e8ae3d1 : SUCCESS in 50m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b0a4d0a2b8e14e99bf61de77d9e6fa2a : SUCCESS in 37m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dfcc4c91c28b42d8aa31b095ee128552 : SUCCESS in 32m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/21382ca997e64e1a9dc950ebd9293220 : FAILURE in 1h 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d627edfeb2903d174e4f3c71187c6862d4fb7e0e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 14:38:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bc8a7d88e94973dc15d21fc0a6eae614094deef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 14:38:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8158a67ea5126c1fd1b95afcd921c54151f4d431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:21:26.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"4b50c19a1af51d0425ace0674f44202a9ee5a279","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-08-14 00:57:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"332e60b4fb90e7ed491921b52fdbad3d30ef6051","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 02:38:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca08f3ca76bf56a4e74bcadc73cb5523f3cebbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 04:52:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22b5556e75624feabbb3a7942cc5f2fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15b490541c040c69924138926c95bbd : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdf6b23edb4c4112af79e847f579c276 : SUCCESS in 10m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a72e1502ae042498fc0dc33275b5c6a : SUCCESS in 9m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e89dc4c2b388480bb015e9ccaf1cb607 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f33bd62f454329971db8f880274742 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f88e2cdb886c4a358c58f24715a7e9cd : SUCCESS in 10m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/95ba41b923354ece9188d9268552bdc8 : SUCCESS in 33m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/984ef96eaf8c4305beb6432148a4cf3a : SUCCESS in 51m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af21e8b9b85c494488c8642407c9a50f : SUCCESS in 51m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/28b9caad6d32479c9dc0c78151ff36eb : SUCCESS in 42m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2a1a22301cfe4aedb88bf9ff16176c9d : SUCCESS in 54m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e7e8cc16dfce469bb9295a56cfb58d14 : FAILURE in 32m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a7909378b2b442749090afc7ef5e9ba4 : SUCCESS in 30m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6d6ee707dc394c419c0aad7ef63058ef : SUCCESS in 2h 06m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"31662e01a87b3cbd3c8bf4a0084ab940ae302699","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 08:00:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4ecbb15a0fce6f1994f56f84b3940f795a3a39ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 09:55:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4054fd1f1f7434b955fc6279d065f5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293ec01a421e450c8d5397cd8a9e999e : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7714f7feafb84aafa766e358cbd798e7 : SUCCESS in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9ab690548cc4ab7a76ea92f77ac5caa : SUCCESS in 8m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bd92b041a844123bf39702b6c26deba : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2ba83ce98541e28e5ce3334b2cf31e : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/569d312305eb480f87bd70879cf6fbd8 : SUCCESS in 10m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9fdac2ff741e45eb82605d18602fa04d : SUCCESS in 44m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82978437c2ee4c72bfc25c654377306a : SUCCESS in 33m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c41321b5c0c4d5e831ddff90873e1f5 : SUCCESS in 47m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/71d128e2ae7b4ae7b0dc30b7c5f6292b : SUCCESS in 47m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3703a80cbcc547b3b40f327fda636566 : SUCCESS in 50m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4bb00645bbff4ddc952b9d7e439fd656 : SUCCESS in 36m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d051b0e1a2914635bd52d1e37cd6d6b6 : SUCCESS in 33m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/73027621f52e48739f9ef67c20a9a7d8 : FAILURE in 1h 49m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c25b494c9dc41c856abaf943a91aa36c1162eb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 09:55:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"45271b294d3b349ebc0ee6f540f28ae303468b7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 11:02:19.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9b0396652754dddb65fd767ace6578b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fdb1e0b1114571bebab747f55ca3d5 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed0a2e470ca14740a3426ab59db827d7 : SUCCESS in 11m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85acb0e07a9048c1965ecb3c708a22e0 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e5c14beb1747c4a2c97c1a6e5442fa : SUCCESS in 7m 27s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/24705aa1ed0b4a30af152f268ea606c1 : SUCCESS in 48m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa5cfaef5b864c0eb9d65ca2bdbd6a83 : FAILURE in 42m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20ed412d638d4c5a964e0de9b7586da0 : SUCCESS in 50m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f52903706c1d4714a4449c85f64f252b : SUCCESS in 52m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cbfab340de9b430490c38ea9b7749b28 : SUCCESS in 1h 02m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b6763982f26146f8956f1a178f0ff414 : SUCCESS in 36m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/685f749d2b8e414187f7ddead90f403e : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":4},{"id":"58ce495973f8d6b4d21b0d7318047e6a2d3ea822","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 12:01:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5742600f418004e2e5f98fb8309b3082a75d7af3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 14:07:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c83f0256478e4136a619c598cba5f69e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d6eeee95d84c17a0b6aa84b9111573 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/567b6620ae964873bfe296cd7d48cc01 : SUCCESS in 8m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56d2cf7bc2904b80abbb11268056324c : SUCCESS in 9m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d31b4e4dbb4e439ba16ee64f732f642a : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f043239a6434b0eb3701b2bd4e42049 : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e9a7c16d964914934601c1caae8e5d : SUCCESS in 10m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b1de4d32321c4333b08b589636db627a : SUCCESS in 49m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26b54653e7d64c099ba32292595e9d86 : SUCCESS in 46m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe6b04c88cfa41d19da9e533197574e0 : SUCCESS in 47m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f6c169f22e84411a8272a3ce652b5c36 : FAILURE in 40m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e031c7d853f949e7bfd8cb70d1aa53a3 : SUCCESS in 46m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5bd8588b7ddb40a69e4719a247fe5a8f : SUCCESS in 35m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ee8271dbf17240aaafd68bc10698d892 : SUCCESS in 32m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1ce3812020294e60a4c8a99dcf869c8f : FAILURE in 1h 58m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7404913b944d2a545657c89ea602954a269b2b0b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 14:24:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a344ffc1fb4f5c5cf6b54a3efb5f17fd055accfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:11:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a73671eb28b4ba19cdf3f295ddada42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89cff5a2e432481eb9197a55795d83d7 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8adf630273344ee282e6c304f8db3dc4 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9cb3a61f0444d2790728ffa190f3016 : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b802a573dd694a148605993532750fb7 : SUCCESS in 13m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9370c58b4f674a008897ffba4456c4d1 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a69a31215d9e44b2a839858650eacfad : SUCCESS in 12m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4d8088b6417e4e94a77aae128fe17223 : SUCCESS in 44m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6828fce35a654071be3d1955b5dc2188 : SUCCESS in 51m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b8881acc4ea479cbd16a2d7217c65cb : SUCCESS in 45m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2ecb9060f11a43148d61f51311fab0f4 : SUCCESS in 57m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3919d9852cdb406e8c74170a0cc4db89 : SUCCESS in 39m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/061ea9159f074e9bbaf0cf1397757bf7 : SUCCESS in 36m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/88dcafab5c2b44629257b3c06bbd25a7 : SUCCESS in 30m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/89c0f759cf08449fbeeab7fbaf9e81a4 : FAILURE in 1h 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03042d81c080a25b8511971b84264d4e91539a3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:11:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f0f2ca618dfc79ff44b8a9260194369a966a508f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 17:00:55.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/282870cef9644acbafd92a9b6de59510\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7e7cf95719420082ac63b1adba3fd5 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71848b3d82d7458aafb321711c9d182a : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f95ce0c14014e96b75fc6514c6bba7b : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d5b0db37b34cf1b8823ab32fd8ce42 : SUCCESS in 8m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/66e491aef96b40e38cac007b70202aa7 : SUCCESS in 48m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a77116c8079b45d7bc2b310a692be78d : SUCCESS in 41m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4730a0764f4b4213b8685ceb76e18a7f : SUCCESS in 46m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4addfd4aa3924d9f9e4daf1f67098b71 : SUCCESS in 39m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bf99d01ee75242cbb846efe70c66b554 : SUCCESS in 48m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f3bca138b82341e9a2ede84a9672df49 : FAILURE in 33m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3c90e1baf8714b63821f707feb1c62f2 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":4},{"id":"0bdad5b70409eabc1ec9c200b4b7a2521346b8c5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 17:43:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8005e3a7966d64f3ca389680a7dd0fdfe9e0877e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:45:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f86182a3cc04cb4b59970e2605b20d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599a5bf612834d4f9e060e71f41d2a64 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13192f1d1296440095a93ff3c1b7cd35 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd1dfa45181e4f11bdb9475457e353e3 : SUCCESS in 9m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/306874fad3a54b9f94d9afe99f475ed3 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d13237f95447b5bc746cda79fe4b6a : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd72b579426f44ae9907fcf3aaa8acb6 : SUCCESS in 10m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/681a1b9ce11b4c939de565a7c7375d53 : SUCCESS in 50m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f978a8f726074b0daf1d86f14eab6454 : SUCCESS in 50m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf4e31474e5348719a750f408c7c335f : SUCCESS in 37m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b89da8e2ab2040a28f879b11bc211c5f : SUCCESS in 50m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9b5c31f44b54412a983c8081ef498380 : SUCCESS in 50m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2896f84895ad4c51b408fc3efe5c7576 : SUCCESS in 33m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bc100670a25e4753aadfc8b3a5342386 : SUCCESS in 33m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/151a209819bf437b9e3c509c7bcefc21 : SUCCESS in 2h 00m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"20059b6ddc9c1b2aa50b20d961614677ce087bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:46:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"14030252417db092a2147fc20be50821601ea54b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 20:34:37.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cf65e8797c0447994ce43f2d4eaf50d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c128134d56428dad58b9ca28dfc4e7 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79021f8e8bd47a3b066d12815237c69 : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab7922a67df0439b96cb3dd7c55ad385 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc497630afe44fc1a81115ad23a1f316 : SUCCESS in 8m 55s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1a5dfbb0dd214f4caf152fc6343c0d47 : SUCCESS in 46m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e872b988b83454c9e65018cc883b736 : SUCCESS in 43m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aac983640f284b79bb859010b3f742d4 : SUCCESS in 44m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/833069f9a5e946eaa0712871c3355878 : SUCCESS in 42m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2ce2d92f25944635a68dc890aecd20a3 : SUCCESS in 45m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8bdc8fdcf7424e268cf4900058412be7 : FAILURE in 40m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8a7ceff2082940a3a91286a10c403ee0 : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":4},{"id":"ce301b4002cf4cfd9b02bdb3c9c96fd8828d8c25","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 23:26:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10b41273b068a1269434e51ae0a9a4af2b0361e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 01:32:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f5a22f61a8946a0bf3a3ec9619faaef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb4952a8ce94189bd47c1aa4a64a594 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3b45e8de4a647c79f617dca2a2b5206 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc95c2301eda48878974d7325b1069c7 : SUCCESS in 9m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5238a1cfe5aa48e189d333bd340ee22c : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2a3e324d1f46e18d8fe26bdb153f10 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d92bfe2a9a4a0f85629bec614bff2c : SUCCESS in 9m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73252a1b271d4f5da3c48b2ee9110350 : SUCCESS in 37m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb057507e44f432ab025b8f2c98e9d6e : SUCCESS in 35m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/271451e03c6148a3b0321b298ad49e27 : SUCCESS in 36m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ab889ebafaaa46118870026d5163ccdb : SUCCESS in 40m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dda9221659cd4120a644c6308a90960a : SUCCESS in 37m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c4e572a46890479a80c851c80da181af : FAILURE in 31m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4f13c2efd5fe47b48d9e155031355960 : SUCCESS in 28m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/115cf3c82e394720b7d5e0c6ce0e10b2 : SUCCESS in 2h 04m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7ffe486bbecfaada1869bd1bb636d666d9c2294","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 13:26:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd4e7d4fef11951a4132045555bf3d9e704c8787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 15:35:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f35911fcf2845b48fbfdc40dc8309e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f3bb605d67f4fa6b8e50a51c0b07ab0 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68fe63e66f824af6bd9824a3070b66a9 : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ee5e3612a4499a93c079027bb11412 : SUCCESS in 10m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a87424f244b145e39d34b98cfaf9127b : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6799d775b5d5470bbacd0cc139934832 : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbf7f3159ebb4a6d9a8ea1dbd1cff6e6 : SUCCESS in 11m 20s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6160fd6dfaa64aa292354617dd9a0430 : SUCCESS in 41m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/905dbc72f39540328443bfbe9f5189bb : SUCCESS in 43m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/023665d2c2fb4047b8cc268e6baa487b : SUCCESS in 43m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5cc878fe57d64c128982fa1535be96e1 : SUCCESS in 47m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4fa03a792a964717aaa5d610c9bcd76f : SUCCESS in 48m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e05fbc4882ae4e22a03f684e92df4ff7 : SUCCESS in 34m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/681bba1225ff44ce8d641d3fa6248517 : SUCCESS in 30m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3bcd60f2548e4405af4e948b7976f602 : TIMED_OUT in 2h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3fd3f9c503141b061e6129b083f5ff1ffcf3b66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 15:36:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fe014808fd5ca4eb4646544689f02cc7f2352b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:36:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c7952bb28f49c1bc22d8ea19c87b31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/958be60979ba4be3b0206ce1ba1be79f : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d295699009ea4801ab5bf66a600e6dcc : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/689a2651f00e438caacb225316f869ae : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8383d5a20748f0a79320a12ad54cf2 : SUCCESS in 7m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0d50ecfa55d14bf5b7f0fdb50ecc037e : SUCCESS in 45m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/788f294c1fc94eba8b02253682087a09 : SUCCESS in 38m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf548d20a1c84cf794f345786c216632 : SUCCESS in 51m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ce777b3138864ca896cef8e403c14c9c : SUCCESS in 42m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/51cafc407029408a9a0311738175a6af : SUCCESS in 54m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/305350db08f74ef6b3d2fbb673c5a6d0 : SUCCESS in 36m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/042210f42d4f4616963fb5582c59800b : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":4},{"id":"59a4df0a0bd892bc838d4f5e8f06768b3c303b64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:36:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"caa76f19ffc8a48be672ba1cf5593c9996504219","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:37:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/247f279d40124dd4a68ddb061199fb45\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26c945f6f04472babcc5a849637c96e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"117f9a794565ef8a3e032824c8c42d421bda7cd0","revisions":{"be7ae7ada48fb793dfb8a8a62f78071f571a6976":{"kind":"REWORK","_number":1,"created":"2021-09-04 16:13:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/807461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/61/807461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/61/807461/1"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 16:11:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 16:12:56.000000000","tz":540},"subject":"Replace deprecated assertDictContainsSubset","message":"Replace deprecated assertDictContainsSubset\n\nThe method is deprecated since Python 3.2[1] and shows the following\nDeprecationWarning.\n\n/usr/lib/python3.9/unittest/case.py:1134: DeprecationWarning:\nassertDictContainsSubset is deprecated\n  warnings.warn(\u0027assertDictContainsSubset is deprecated\u0027,\n\n[1] https://docs.python.org/3/whatsnew/3.2.html#unittest\n\nCloses-Bug: #1938103\nChange-Id: I17f56606eb43fba46281e9bdc08ace8727455a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be7ae7ada48fb793dfb8a8a62f78071f571a6976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be7ae7ada48fb793dfb8a8a62f78071f571a6976"}]},"branch":"refs/heads/master"},"f96f7572c6e635dcea902bc504f78ad0f6327781":{"kind":"REWORK","_number":2,"created":"2022-08-12 09:48:30.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/61/807461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/61/807461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/61/807461/2"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 16:11:45.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-12 09:48:30.000000000","tz":0},"subject":"Replace deprecated assertDictContainsSubset","message":"Replace deprecated assertDictContainsSubset\n\nThe method is deprecated since Python 3.2[1] and shows the following\nDeprecationWarning.\n\n/usr/lib/python3.9/unittest/case.py:1134: DeprecationWarning:\nassertDictContainsSubset is deprecated\n  warnings.warn(\u0027assertDictContainsSubset is deprecated\u0027,\n\n[1] https://docs.python.org/3/whatsnew/3.2.html#unittest\n\nCloses-Bug: #1938103\nChange-Id: I17f56606eb43fba46281e9bdc08ace8727455a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f96f7572c6e635dcea902bc504f78ad0f6327781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f96f7572c6e635dcea902bc504f78ad0f6327781"}]},"branch":"refs/heads/master"},"1db771c02b95ee57d654c8ce73b09912169765bc":{"kind":"REWORK","_number":3,"created":"2022-08-12 09:58:10.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/61/807461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/61/807461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/61/807461/3"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 16:11:45.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-12 09:58:09.000000000","tz":0},"subject":"Replace deprecated assertDictContainsSubset","message":"Replace deprecated assertDictContainsSubset\n\nThe method is deprecated since Python 3.2[1] and shows the following\nDeprecationWarning.\n\n/usr/lib/python3.9/unittest/case.py:1134: DeprecationWarning:\nassertDictContainsSubset is deprecated\n  warnings.warn(\u0027assertDictContainsSubset is deprecated\u0027,\n\n[1] https://docs.python.org/3/whatsnew/3.2.html#unittest\n\nCloses-Bug: #1938103\nChange-Id: I17f56606eb43fba46281e9bdc08ace8727455a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1db771c02b95ee57d654c8ce73b09912169765bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1db771c02b95ee57d654c8ce73b09912169765bc"}]},"branch":"refs/heads/master"},"117f9a794565ef8a3e032824c8c42d421bda7cd0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-08-12 11:59:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/807461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/61/807461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/61/807461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/61/807461/4"}}},"commit":{"parents":[{"commit":"88c7a5d1ff7c0abafa1fd776dff0ce14a4241ca9","subject":"Update python testing as per zed cycle testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/88c7a5d1ff7c0abafa1fd776dff0ce14a4241ca9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 16:11:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-12 11:59:12.000000000","tz":0},"subject":"Replace deprecated assertDictContainsSubset","message":"Replace deprecated assertDictContainsSubset\n\nThe method is deprecated since Python 3.2[1] and shows the following\nDeprecationWarning.\n\n/usr/lib/python3.9/unittest/case.py:1134: DeprecationWarning:\nassertDictContainsSubset is deprecated\n  warnings.warn(\u0027assertDictContainsSubset is deprecated\u0027,\n\n[1] https://docs.python.org/3/whatsnew/3.2.html#unittest\n\nCloses-Bug: #1938103\nChange-Id: I17f56606eb43fba46281e9bdc08ace8727455a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/117f9a794565ef8a3e032824c8c42d421bda7cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/117f9a794565ef8a3e032824c8c42d421bda7cd0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
