)]}'
{"id":"openstack%2Fmistral~809373","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I51eab0503c101f449cbe322ca570ee94230058e8","project":"openstack/mistral","branch":"stable/train","topic":"cleanup-c7-jobs-stable/train","hashtags":[],"change_id":"I51eab0503c101f449cbe322ca570ee94230058e8","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-16 12:12:37.000000000","updated":"2023-12-21 14:17:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809412,"cherry_pick_of_patch_set":1,"meta_rev_id":"5e87252f3f4232c8dfdc6d3dea0f04612283a631","_number":809373,"virtual_id_number":809373,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-16 13:08:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 13:08:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e81ee39821be3da15025cf244bfcf263d489fe83","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 12:12:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"0818b567a5023ada9294f4ea22c56da3d17a033d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 12:13:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7ad85cfb784ace3b490dd3b009d9bdc5fa2eefbd","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 12:13:41.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"624498859bfb42d0b82005977b1962ce22f0cf76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:08:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ffdeb84a5640ce9c15baf360d5bcf4 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/316a9f06497542ceab4b846c0a4fd97c : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d03c104150a3423aadbe17c9f92a132c : SUCCESS in 11m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d308063786584dbb87d11b03892c55bb : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c9d3d6a2214db4be3c8faecf37cae3 : SUCCESS in 7m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/21edeb6ba90145ac9ca78426fb93f659 : SUCCESS in 36m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/272d9b87a3c247ed8dbfa4bae2b22ffa : SUCCESS in 46m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7ce2bacf3882417b85a08b953965b28a : SUCCESS in 40m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a0410ad8c6074b5dbbbce958b17e0bad : FAILURE in 29m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/315c5097f06c48de9fce7e6cf8db547f : SUCCESS in 30m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c412aac440994a1695c41a214760cc6c : FAILURE in 36m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/918fa69a0f474dd888e872b88886c3e1 : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5e87252f3f4232c8dfdc6d3dea0f04612283a631","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:17:03.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87a5a2d0f6b515bbd2cc593c68456f7879937224","revisions":{"2ef33af470d9e7f443ec900ee17f8e6b549d2f96":{"kind":"REWORK","_number":1,"created":"2021-09-16 12:12:37.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/73/809373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/809373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/809373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/809373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/809373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/809373/1"}}},"commit":{"parents":[{"commit":"62b0b37e3a01ed12f6791d596a93babc698e7217","subject":"Merge \"Update requirements to include oauthlib\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/62b0b37e3a01ed12f6791d596a93babc698e7217"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 12:11:06.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 12:12:37.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those c7 jobs from templates/layouts.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I51eab0503c101f449cbe322ca570ee94230058e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ef33af470d9e7f443ec900ee17f8e6b549d2f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ef33af470d9e7f443ec900ee17f8e6b549d2f96"}]},"branch":"refs/heads/stable/train"},"87a5a2d0f6b515bbd2cc593c68456f7879937224":{"kind":"REWORK","_number":2,"created":"2021-09-16 12:13:37.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/73/809373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/809373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/809373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/809373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/809373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/809373/2"}}},"commit":{"parents":[{"commit":"62b0b37e3a01ed12f6791d596a93babc698e7217","subject":"Merge \"Update requirements to include oauthlib\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/62b0b37e3a01ed12f6791d596a93babc698e7217"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 12:11:06.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 12:13:34.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those c7 jobs from templates/layouts.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I51eab0503c101f449cbe322ca570ee94230058e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87a5a2d0f6b515bbd2cc593c68456f7879937224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87a5a2d0f6b515bbd2cc593c68456f7879937224"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
