)]}'
{"id":"openstack%2Fmistral~809375","triplet_id":"openstack%2Fmistral~stable%2Fpike~Icabf2bfcda8cecf3fe4e5d19b7376af92a2ab7d3","project":"openstack/mistral","branch":"stable/pike","topic":"cleanup-featuresets-from-old-branches-stable/pike","hashtags":[],"change_id":"Icabf2bfcda8cecf3fe4e5d19b7376af92a2ab7d3","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-16 13:24:57.000000000","updated":"2022-10-14 17:31:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809397,"cherry_pick_of_patch_set":1,"meta_rev_id":"9890ddc778fbbd93d1a3a30c63a3da11f6c94d24","_number":809375,"virtual_id_number":809375,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-16 14:15:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 13:25:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c24a7de94a4d7e312ebba0c1ddb33f01b1e0b42","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 13:24:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"220132ae6945eb7d008fbef33a97992d07a8b104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:25:21.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/mistral on branch stable/pike.  The error was:\n\n  while scanning a simple key\n  in \"openstack/mistral/.zuul.yaml@stable/pike\", line 107, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/mistral/.zuul.yaml@stable/pike\", line 108, column 33","accounts_in_message":[],"_revision_number":1},{"id":"44b48f6f3ed5e7ffa7f2b3556399d3ccadde0f81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 13:28:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6161d604f0bde7b8bc3c63d3634fc0fbe99b1a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:28:50.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/mistral on branch stable/pike.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/mistral/.zuul.yaml@stable/pike\", line 78, column 38","accounts_in_message":[],"_revision_number":2},{"id":"f38bea80ebc43ce2ac8a77326d51e3cc54564b6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 13:30:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"16c8042aa15b719b30962ef26bb02bc7ffa51fbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 13:31:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f3267985700c472a70075e3612bd067e317aa22d","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 13:31:31.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"5cb07fd2bb39e775a1739ee395169c1224284bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 14:15:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73722343e1ba42bd92b352598d638c02 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82ac4f67eb7e4598b4c74311c83d8224 : RETRY_LIMIT in 2m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8cef163d6f744b62b9a6594ff190f351 : RETRY_LIMIT in 3m 27s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9f356c0ad6954e7aaf2270d8b00425b7 : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71e2af4cec7a47259062e1a3a836a7cc : RETRY_LIMIT in 3m 43s (non-voting)\n- rally-dsvm-mistral-task https://zuul.opendev.org/t/openstack/build/a8bbc01d177e4c06a4bd628c40aa5fc1 : FAILURE in 34m 11s (non-voting)\n- mistral-devstack-dsvm https://zuul.opendev.org/t/openstack/build/5ac88bb6b8ec4e878046bfbd19ed7a78 : FAILURE in 31m 07s (non-voting)\n- mistral-devstack-dsvm-non-apache https://zuul.opendev.org/t/openstack/build/2084373806d94f60a539819fd3ffdf29 : FAILURE in 37m 22s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3a8946b7e7fe43c1818d11b7b86bef38 : FAILURE in 6m 01s (non-voting)\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d38d0b02241f455f88f804fbec9fab4c : FAILURE in 5m 01s (non-voting)\n- mistral-devstack-dsvm-kombu https://zuul.opendev.org/t/openstack/build/007cdb4424ed46409346a2bf8479fb29 : FAILURE in 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9890ddc778fbbd93d1a3a30c63a3da11f6c94d24","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-14 17:31:13.000000000","message":"Abandoned\n\nstable/pike branch of mistral has been EOL\u0027d, patches need to be abandoned to be able to delete the branch.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fa0f643d32c976b1fb6fccd97e6688a610d7ec8b","revisions":{"def5a2ca09f4f827f6f61fb9f724b40ee8269a21":{"kind":"REWORK","_number":1,"created":"2021-09-16 13:24:57.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/75/809375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/75/809375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/75/809375/1"}}},"commit":{"parents":[{"commit":"4b078bc581b450149c46e063529baa4ac38a1b7f","subject":"Process all task batches in wf output evaluation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b078bc581b450149c46e063529baa4ac38a1b7f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 10:59:51.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 13:24:57.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Icabf2bfcda8cecf3fe4e5d19b7376af92a2ab7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/def5a2ca09f4f827f6f61fb9f724b40ee8269a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/def5a2ca09f4f827f6f61fb9f724b40ee8269a21"}]},"branch":"refs/heads/stable/pike"},"e4743dac4fb5421cbd8da87633740e121fed9d15":{"kind":"REWORK","_number":2,"created":"2021-09-16 13:28:14.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/75/809375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/75/809375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/75/809375/2"}}},"commit":{"parents":[{"commit":"4b078bc581b450149c46e063529baa4ac38a1b7f","subject":"Process all task batches in wf output evaluation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b078bc581b450149c46e063529baa4ac38a1b7f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 10:59:51.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 13:28:06.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Icabf2bfcda8cecf3fe4e5d19b7376af92a2ab7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4743dac4fb5421cbd8da87633740e121fed9d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4743dac4fb5421cbd8da87633740e121fed9d15"}]},"branch":"refs/heads/stable/pike"},"b738bc6e68f1295a6aa8a53e5521ad74c17f77f2":{"kind":"REWORK","_number":3,"created":"2021-09-16 13:30:56.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/75/809375/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/75/809375/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/75/809375/3"}}},"commit":{"parents":[{"commit":"4b078bc581b450149c46e063529baa4ac38a1b7f","subject":"Process all task batches in wf output evaluation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b078bc581b450149c46e063529baa4ac38a1b7f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 10:59:51.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 13:30:52.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Icabf2bfcda8cecf3fe4e5d19b7376af92a2ab7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b738bc6e68f1295a6aa8a53e5521ad74c17f77f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b738bc6e68f1295a6aa8a53e5521ad74c17f77f2"}]},"branch":"refs/heads/stable/pike"},"fa0f643d32c976b1fb6fccd97e6688a610d7ec8b":{"kind":"REWORK","_number":4,"created":"2021-09-16 13:31:23.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/75/809375/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/75/809375/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/809375/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/75/809375/4"}}},"commit":{"parents":[{"commit":"4b078bc581b450149c46e063529baa4ac38a1b7f","subject":"Process all task batches in wf output evaluation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b078bc581b450149c46e063529baa4ac38a1b7f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 10:59:51.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 13:31:19.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Icabf2bfcda8cecf3fe4e5d19b7376af92a2ab7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa0f643d32c976b1fb6fccd97e6688a610d7ec8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa0f643d32c976b1fb6fccd97e6688a610d7ec8b"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
